)]}'
{"id":"openstack%2Fopenstack-ansible-os_ironic~732274","triplet_id":"openstack%2Fopenstack-ansible-os_ironic~stable%2Fussuri~I5e90e6dbc9f80043d661199dc7f864e1910749f9","project":"openstack/openstack-ansible-os_ironic","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I5e90e6dbc9f80043d661199dc7f864e1910749f9","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-06-01 13:14:57.000000000","updated":"2020-06-01 18:33:06.000000000","submitted":"2020-06-01 18:31:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732274-1591036296233-f06ad413","meta_rev_id":"ded965b448d13004ed075e72e37efd1d52783082","_number":732274,"virtual_id_number":732274,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-01 18:31:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-06-01 18:18:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-06-01 18:18:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-01 18:18:47.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-06-01 18:31:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe6650bc9152b04b5010597de34045d85792fcbc","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-06-01 13:14:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4ed6d36417f039f548c76febeaeb72c641b7662f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 13:45:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/84c91f624a004a96ae5cc87f059e920a : SUCCESS in 9m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31b4898fbb7b4b0482085e5c86b566e6 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6036844f9ea4baab80d81da56437e55 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":1},{"id":"875bef3744ec271882fb740f6ddbc484baad00dd","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-01 18:18:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1a25e3f95ed3ae72d3cc5bbe24276e2c85332435","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:18:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"380ee57fb6dbd100e60c65e25c1d74650544f678","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:31:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fd36d68575b64d9db92633e2b4cf8965 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35645b78b85d41f4b351979959f1d6b7 : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/be96ff7d47584fee962d5a4eba5fa901 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":1},{"id":"8d6fd7a2cb288ca2468dc49b5243e484fabcbe85","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:31:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ecfaff1a85b480428586746471d59d95cc20a408","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-01 18:33:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/209c78c98f4247cca511c58c6e3afb63 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeb14e25fdca45d0a8a81145a757b9ea : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bae9178669d83943c6f6fc21c5805997247fbb33","revisions":{"bae9178669d83943c6f6fc21c5805997247fbb33":{"kind":"REWORK","_number":1,"created":"2020-06-01 13:14:57.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/74/732274/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/74/732274/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/74/732274/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/74/732274/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/74/732274/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/74/732274/1"}}},"commit":{"parents":[{"commit":"31fded6bee978e482278c7392412883f7921b055","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/31fded6bee978e482278c7392412883f7921b055"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-06-01 13:14:55.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-06-01 13:14:55.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I5e90e6dbc9f80043d661199dc7f864e1910749f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/bae9178669d83943c6f6fc21c5805997247fbb33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/bae9178669d83943c6f6fc21c5805997247fbb33"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
