)]}'
{"id":"openstack%2Fopenstack-ansible-os_ironic~772284","triplet_id":"openstack%2Fopenstack-ansible-os_ironic~master~I06958da8881c93abde25c114d06943d777ea2b0a","project":"openstack/openstack-ansible-os_ironic","branch":"master","topic":"osa-new-pip","hashtags":[],"change_id":"I06958da8881c93abde25c114d06943d777ea2b0a","subject":"Move ironic pip packages from constraints to requirements","status":"MERGED","created":"2021-01-25 10:06:52.000000000","updated":"2021-02-15 18:55:25.000000000","submitted":"2021-02-15 18:53:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"772284-osa-new-pip","meta_rev_id":"ccd0a59ab690094ff2e2c1f0d3b8f19408aa7575","_number":772284,"virtual_id_number":772284,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-15 18:53:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-02-15 08:51:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-02-15 17:17:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2021-02-15 17:17:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-01-25 13:25:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-15 08:51:55.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2021-02-15 17:17:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"}],"messages":[{"id":"48deff3de9c6664a098b23585fa877df0b3ad17c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-25 10:06:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1138b272d53b16c2c13379a8c673eae73b5352a8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-25 13:25:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/de53507ac6494e408822a76370051542 : SUCCESS in 10m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/16e049f7a0dc4136ad09b263c85d8d25 : FAILURE in 1h 15m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d3a28ba20175445e8902494784aabf0e : FAILURE in 1h 22m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7e7fbe63d35641a49496d3869b543862 : FAILURE in 1h 20m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ebead972cd664f36a5ce23f42e83fc2c : FAILURE in 1h 17m 30s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/840234736e224d2a9182bd86d89e76b8 : FAILURE in 2h 22m 01s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6541bef28daf41f6aa39eaf5f10aace0 : FAILURE in 1h 43m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f568e49247b4b70811b84e5a9962696 : SUCCESS in 5m 48s","accounts_in_message":[],"_revision_number":1},{"id":"ff24b5662dfd58716bb6742eca88a5efbef78db7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-26 08:51:10.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"33ad9f5b392bc49211f85ab2809acf5e4c74d0ad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 10:52:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/152b8bdcbfef47bc85a317723f1a2e45 : SUCCESS in 11m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/99c6100d081a433f8c8bfa7f1cd48ff2 : FAILURE in 1h 07m 20s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/3eab731cb9f041aa8748c8ac5377504f : FAILURE in 1h 42m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cae9a087cc79403ba60de8a3dbb8e4c1 : FAILURE in 1h 54m 50s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e2bffc12a0ee497194ed73611b590f4d : FAILURE in 1h 10m 42s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a04d8ed0ed5b457e8dcd31b5fa14944a : FAILURE in 1h 54m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/01dde9358c59476fa51372ea3755c890 : FAILURE in 1h 41m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e23e31dbc89644dca91068b9b5a649b8 : SUCCESS in 6m 37s","accounts_in_message":[],"_revision_number":2},{"id":"b44b7d6eb22baff9abb6d41b2f142f8b6270ad1c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-26 12:58:50.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a939d83564af9815eb9fbabe39152e260173cc33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-26 16:14:20.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0527542e428e4bac8231f39f3f34787b : SUCCESS in 11m 45s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/7b1728a7924c4cf8917ae20ab12dbee7 : FAILURE in 1h 35m 34s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2b9279b905194e0c84b223bd060c4b54 : FAILURE in 1h 24m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d16fc0e8aa84117b1e6aa9bb6e17fd2 : FAILURE in 1h 22m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d99fe48b2a5c413c908fb8dc032db14e : FAILURE in 1h 09m 00s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6798112894224af79b39bcd34474f5a9 : TIMED_OUT in 3h 03m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/459520b314094904b5e8f7cec2bff62f : FAILURE in 1h 52m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a8b878f929c40359cc1c86ae702fd9b : SUCCESS in 11m 58s","accounts_in_message":[],"_revision_number":2},{"id":"d13792a5421aaf438ae1548f84dd5aee40b6b3a8","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-26 21:32:23.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"27aacae14b4632acb01676412df550c66ee537c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 00:02:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3000510e7a3648d2819fa7463863053a : SUCCESS in 8m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bea3fb5f53d34d96b27b0ac3de9f0b45 : FAILURE in 1h 08m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/087ced09f1f7435d8749a51b180c30ea : FAILURE in 1h 24m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5b793b07e4364816a371ce20cbdb2073 : FAILURE in 1h 27m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/58759a8becb7428993735bfda9f5e0c2 : FAILURE in 1h 11m 21s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1d530b98ae074c5caef7da27588424a8 : FAILURE in 2h 19m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c8b21b57af034432b66b9dab947c95ee : FAILURE in 1h 41m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3817ad0d3ac499cb8c777d36a17c262 : SUCCESS in 5m 40s","accounts_in_message":[],"_revision_number":2},{"id":"8d6fbd0f7464f16bcad3e8d391a9bfb9e03549e1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-27 09:36:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"c785fe6e431a1869a57ccec1537e144c4ac32f13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 12:06:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/8c8916bbce4e410d91f453664f642d34 : SUCCESS in 8m 11s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3a0b8567676943cebb53d0ecbbd5e289 : FAILURE in 1h 09m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/97dcaa4185f143be9623ec653b505383 : FAILURE in 1h 22m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f72992b61e1743c4bba710afff159970 : FAILURE in 1h 24m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7354f20d1f8c40e3b2753a4fa2c81c4d : FAILURE in 1h 04m 50s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/61e0aa2ade2043018057803edb89712f : FAILURE in 2h 20m 16s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6ac686e96bee4b74b695609046dacf37 : FAILURE in 1h 42m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/404ee8f5dda04f5ab7a6768a8819ec6f : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":2},{"id":"4af9e14133f786a665f0a68805c0131d3fe52cbe","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-27 12:31:16.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"91cda596a46b57df2ad9d72de29e5516fc25a619","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-27 15:02:59.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0f0bed7f36c74cf5969a34944556851b : SUCCESS in 12m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a44e61e45054428fb59e3be424478791 : FAILURE in 1h 02m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/83ea7e7647d84bdc98274ce220b69368 : FAILURE in 1h 19m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4ea3058bb8714d35b2099e331204651e : FAILURE in 1h 20m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bff4711a16b242fb83185e47c0be3643 : FAILURE in 1h 04m 04s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4f01262d2add42f0bdbd69038c1db21a : FAILURE in 2h 19m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9093836fc6b14a34bb8c0ee42ac24d4e : FAILURE in 1h 43m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/517828d541ca45b2be4507ee3294901f : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":2},{"id":"01a3eea3010782e0a8f911c6761d257e1c41e2ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-28 16:24:35.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"bb815d16da756d2fb863679df86f7fc9c6618a0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-28 16:36:20.000000000","message":"Patch Set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"16d9fe5447146ab6dee55f39e7a44d5415c6609b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-28 19:07:15.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dbe9164fc5d74421800bc28f2f181e5d : SUCCESS in 9m 43s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/47a0528e56784e028c5a43b4ec28eec7 : SUCCESS in 1h 11m 42s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b5c8ecba1eac4cc387222b3bacf4efaa : SUCCESS in 1h 24m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4e62589a6b0e48f18d5e1d451358a8a0 : SUCCESS in 1h 29m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0efec24e275142f5bf1543ad12ba68ad : SUCCESS in 1h 10m 17s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9bcd76b88ab442499027d72f0b7472a5 : FAILURE in 2h 03m 12s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/8ff5b691a1ef4e4384011c02255ba3a5 : FAILURE in 1h 40m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6fbcce6f6e454cc293af1c6b177a43a5 : SUCCESS in 8m 27s","accounts_in_message":[],"_revision_number":4},{"id":"005123ed49d6996305f4e0fb746a1c6a0647414a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-28 22:23:16.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"90a249b1b8dcb8ec4d0aef56b4894b322cabe388","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-29 00:08:24.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/44bb091f54b2414caf44708881e7460d : SUCCESS in 11m 21s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a4362665f4b94d15b9199efc74a0175e : SUCCESS in 1h 19m 03s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/8d3de0a2c1234572bc0f23948dd23ccf : SUCCESS in 1h 25m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/df8666bbe3aa400fba70ad241731e6bf : SUCCESS in 1h 36m 27s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/733e748b3b5f4022b5bbe47266389ff0 : SUCCESS in 1h 20m 56s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e99abc38ad7c495b80de8abe2cc1c925 : FAILURE in 18m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/012da7292237478ea4486bea2a80ab42 : FAILURE in 20m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d56d9dc9f69341b889227387c1402dc4 : SUCCESS in 12m 04s","accounts_in_message":[],"_revision_number":4},{"id":"d6d7aa507a114101f909d6e64839d54875ee700e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-01-29 06:47:33.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"80a757329f9756fe80e6e453c25986cafc294b19","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-05 17:21:35.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"c64090f69783f9a7ffaefda3ad66f618ca76e58e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 19:33:11.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e3a2f3e2bb5d453b8af17263bca7405a : SUCCESS in 12m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1a2d798c32354b009b2a2d3305849ba0 : SUCCESS in 1h 10m 33s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e6ea2fea0fb44cffa80d5266d19f8101 : SUCCESS in 1h 27m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/82f50ceeb42142c49a67aa190fd7667c : SUCCESS in 1h 19m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/724eacc69407415481c5ca5194970ada : SUCCESS in 1h 13m 01s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/86493164c2164e6daf4f7bb50b81e1a6 : FAILURE in 1h 59m 37s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/da78882f266b49dc9888c6faa28f068b : FAILURE in 1h 40m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f5971d445df4dca9cac01530e512224 : SUCCESS in 7m 47s","accounts_in_message":[],"_revision_number":4},{"id":"6432d1fe24e0cc5563f747abd96f1fbb636442e6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-14 08:15:11.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"f10e7308afd5960da8c5e471d5d76f4b8f41d5bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-14 10:15:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/7688e1442ed14b40baa45dd15ab7c7be : SUCCESS in 8m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1bf473bc5c1446bc982b8fa0b1cea4c0 : SUCCESS in 1h 00m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/cc7d02def20545f4b9ff3f6814af35e6 : SUCCESS in 1h 12m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c6c72f891569464db88db0239e71784e : SUCCESS in 1h 09m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b51aaa5adb3a49788b4c1ba1ff3accda : SUCCESS in 58m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/5ba71dcb37b84ef5a1a6ee60db2e4291 : SUCCESS in 1h 57m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4a27e1170894f8396077ea5b1aab7c9 : SUCCESS in 1h 30m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec03aee155b461fa1471678642bb1fa : SUCCESS in 4m 36s","accounts_in_message":[],"_revision_number":4},{"id":"518cca0285f29d8bf9a01c57d2b4bf039d80c446","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-15 08:51:55.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"4fceb66cd721ab7d3378581c7746c5c301699d02","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-15 17:17:56.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"66ee3d9c9b93f7d1e5757d85fd456081629adb02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 17:18:12.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"95b3cb7d05f360c4890b29c8e1a22b84131b2011","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 18:53:09.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e5a3a77759bf404e8a98b268e8c8e986 : SUCCESS in 10m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b698c326fb174d72bee110659def1ff2 : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1323b678a610480bb4c7b5d17eadc11a : SUCCESS in 1h 01m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ad5b55013d84e23850824bb9c807b51 : SUCCESS in 1h 21m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e25318bf2a664cc2b8fcb6cf1a42a158 : SUCCESS in 1h 07m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc27a5368c14a2eaec1d4293ccd5404 : SUCCESS in 5m 51s","accounts_in_message":[],"_revision_number":4},{"id":"967b2a8eba43ec96c87221525191f034f01b289e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 18:53:10.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"ccd0a59ab690094ff2e2c1f0d3b8f19408aa7575","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-15 18:55:25.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d295b4fa50f40d1a6e29cc1eb620b54 : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b4df743cfe82fab40835d3435a46f909e659c742","revisions":{"c89b711a451d053eea9fd18d9af187bf2d1b073a":{"kind":"REWORK","_number":1,"created":"2021-01-25 10:06:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/84/772284/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/84/772284/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/1"}}},"commit":{"parents":[{"commit":"67733c8f0cb13c467eb10f256a93d878c56d4743","subject":"Define condition for the first play host one time","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/67733c8f0cb13c467eb10f256a93d878c56d4743"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:06:15.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:06:15.000000000","tz":0},"subject":"Move ironic pip packages from constraints to requirements","message":"Move ironic pip packages from constraints to requirements\n\nThis is necessary to use the new pip resolver\n\nChange-Id: I06958da8881c93abde25c114d06943d777ea2b0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/c89b711a451d053eea9fd18d9af187bf2d1b073a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/c89b711a451d053eea9fd18d9af187bf2d1b073a"}]},"branch":"refs/heads/master"},"f5f2b2f8e3118246369530aec21a4cf1b45e7c69":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-01-26 08:51:10.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/84/772284/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/84/772284/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/2"}}},"commit":{"parents":[{"commit":"67733c8f0cb13c467eb10f256a93d878c56d4743","subject":"Define condition for the first play host one time","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/67733c8f0cb13c467eb10f256a93d878c56d4743"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:06:15.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-26 08:51:10.000000000","tz":0},"subject":"Move ironic pip packages from constraints to requirements","message":"Move ironic pip packages from constraints to requirements\n\nThis is necessary to use the new pip resolver\n\nDepends-On: Ia89ee301edc4b4a0bc2063e0cb242e510c02a26a\nChange-Id: I06958da8881c93abde25c114d06943d777ea2b0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/f5f2b2f8e3118246369530aec21a4cf1b45e7c69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/f5f2b2f8e3118246369530aec21a4cf1b45e7c69"}]},"branch":"refs/heads/master","description":"Edit commit message"},"911c845e0184873637c9a527fe8b22b550f416bb":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-01-28 16:24:35.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/84/772284/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/84/772284/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/3"}}},"commit":{"parents":[{"commit":"67733c8f0cb13c467eb10f256a93d878c56d4743","subject":"Define condition for the first play host one time","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/67733c8f0cb13c467eb10f256a93d878c56d4743"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:06:15.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-28 16:24:35.000000000","tz":0},"subject":"Move ironic pip packages from constraints to requirements","message":"Move ironic pip packages from constraints to requirements\n\nThis is necessary to use the new pip resolver\n\nDepends-On: I6003a1004b6042a804ea142684b42b17f16d6da8\nDepends-On: Ia89ee301edc4b4a0bc2063e0cb242e510c02a26a\nChange-Id: I06958da8881c93abde25c114d06943d777ea2b0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/911c845e0184873637c9a527fe8b22b550f416bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/911c845e0184873637c9a527fe8b22b550f416bb"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b4df743cfe82fab40835d3435a46f909e659c742":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2021-01-28 16:36:20.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/84/772284/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/84/772284/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/84/772284/4"}}},"commit":{"parents":[{"commit":"67733c8f0cb13c467eb10f256a93d878c56d4743","subject":"Define condition for the first play host one time","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/67733c8f0cb13c467eb10f256a93d878c56d4743"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-25 10:06:15.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-01-28 16:36:20.000000000","tz":0},"subject":"Move ironic pip packages from constraints to requirements","message":"Move ironic pip packages from constraints to requirements\n\nThis is necessary to use the new pip resolver\n\nDepends-On: I6003a1004b6042a804ea142684b42b17f16d6da8\nChange-Id: I06958da8881c93abde25c114d06943d777ea2b0a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/b4df743cfe82fab40835d3435a46f909e659c742"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/b4df743cfe82fab40835d3435a46f909e659c742"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
