)]}'
{"id":"openstack%2Fopenstack-ansible-os_ironic~891741","triplet_id":"openstack%2Fopenstack-ansible-os_ironic~stable%2F2023.1~I3952a4e5514824381410d87ed6d535f13ec40498","project":"openstack/openstack-ansible-os_ironic","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2023-08-22 09:49:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3952a4e5514824381410d87ed6d535f13ec40498","subject":"Fix lack of constraints for ironic-inspector","status":"MERGED","created":"2023-08-17 13:30:19.000000000","updated":"2023-08-22 09:50:01.000000000","submitted":"2023-08-22 09:49:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"891741","cherry_pick_of_change":890335,"cherry_pick_of_patch_set":1,"meta_rev_id":"dd856f45047c4a029e8bd68685ebb29ef0d632ed","_number":891741,"virtual_id_number":891741,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-08-22 09:49:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2023-08-22 07:30:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2023-08-21 19:47:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2023-08-22 07:30:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-17 13:30:19.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"CC"},{"updated":"2023-08-17 16:40:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-08-21 13:21:59.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2023-08-21 19:47:51.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2023-08-22 07:30:53.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"d82cf35bc6bea88392a3ae8f341df943f2218a10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2023-08-17 13:30:19.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"962c35c4ac85a2c3ce73a64f29b3b1a9150cd667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-17 16:40:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a5b706f067934ecfbb90dc3cb1204d3c\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fb938cb71c5c4dab84c362cb772b0e42 : SUCCESS in 12m 00s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/110a301ca225482cacde0926d8307871 : SUCCESS in 1h 54m 10s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/9dfcfab3051b467f8df3a28412a221c2 : SUCCESS in 1h 46m 00s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/b40bff94491747e49e973fcb5652d650 : SUCCESS in 1h 46m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6c091d811e20460b9498ea1e2c7652cf : SUCCESS in 1h 50m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ce50d61e08744edfb4590e37883581a1 : SUCCESS in 1h 37m 04s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/1a89f6c7edbb4e5595bed4cdb8629848 : SUCCESS in 2h 05m 14s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/73ba2ae4730e4cc1937ffa67da317dd8 : SUCCESS in 1h 49m 21s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/4675cd33a37744cfa2c6592953f7ebb1 : DISK_FULL in 2h 35m 01s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/017b3f16fc804fb9b6f34ddeb51d353d : SUCCESS in 2h 27m 41s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ce21d9097e544b2cbab2b5070ad75e1d : FAILURE in 2h 07m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/767eecd8ef834cf5bcd920281d3f8eac : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f839775968a6ef05a776e5d5e02e0178ccfec47d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-21 13:21:59.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8ea3596a872fda3c06bfa73502d5a78f1ce5ca9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-21 16:20:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/471226332ec7405697756eb0f5af37d6\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/90417b7cd28a4fd68ce581fe4072aa59 : SUCCESS in 17m 37s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/4f64b33739b345d89631c1963a4e80b2 : SUCCESS in 1h 46m 19s\n- openstack-ansible-deploy-aio_metal-centos-9-stream https://zuul.opendev.org/t/openstack/build/728e42198d0440c09c9dcdfd4b9108cd : SUCCESS in 1h 58m 52s (non-voting)\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/79b695eaef7c4674affcc624a46d7043 : SUCCESS in 1h 39m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7d9a7d9851a84c92b07c234714e81871 : SUCCESS in 2h 01m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7fbe71c693ab4210a125f23a999cceff : SUCCESS in 1h 34m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/664e0b1034aa4016b5d4ad102bdd9dd9 : SUCCESS in 2h 08m 13s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/34ccbcaef36340349d50d8a956580178 : SUCCESS in 1h 40m 25s\n- openstack-ansible-upgrade-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/ce71bc43c2eb474789953d1bd0b232ab : SUCCESS in 2h 26m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7490a3e281154fe3a8ed8e37deaa1ae0 : SUCCESS in 2h 42m 09s\n- openstack-ansible-upgrade_yoga-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/14f93084b1eb407ca20624c2c2919d78 : SUCCESS in 2h 22m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db4ccbc5457040f49fd3ef5ac726d1cd : SUCCESS in 4m 42s","accounts_in_message":[],"_revision_number":1},{"id":"6c9d23d09e93a97a5c976a30e2e252d01c7b4b48","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2023-08-21 19:47:51.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e7b5814d679e00f23b9ce160f336ac70655b139f","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2023-08-22 07:30:53.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"77986ad5a19508c55c6f45ed746e19f0f3895080","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 07:32:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32188676208b523a47601e1256e2099d6642299d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 09:49:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb0c9155857b409b95a52eb3de59f35e\n\n- openstack-ansible-linters-ubuntu-focal https://zuul.opendev.org/t/openstack/build/072a5e3d2d4e492abcfcd71589f32aaf : SUCCESS in 11m 55s\n- openstack-ansible-deploy-aio_metal-debian-bullseye https://zuul.opendev.org/t/openstack/build/ef1645e9cace42cead8263d5b3d6827f : SUCCESS in 1h 21m 04s\n- openstack-ansible-deploy-aio_metal-rockylinux-9 https://zuul.opendev.org/t/openstack/build/564fa29599e7440c96a14018c8df1348 : SUCCESS in 1h 45m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f4af34b3849f491690403d6780713b5b : SUCCESS in 1h 41m 41s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-9 https://zuul.opendev.org/t/openstack/build/09d18e3efe104a0fab5f496eb12f726f : SUCCESS in 2h 09m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b1609035dfef443996f0aecc9b7faeaa : SUCCESS in 1h 46m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f4da3edd163c4ab793aeed4e30bcadc8 : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":1},{"id":"75ddec1cf6885008eb3708bb780f95dcd483c23d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 09:49:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dd856f45047c4a029e8bd68685ebb29ef0d632ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 09:50:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d48638b27c08494193aad0b2db5e0082\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dfd62778d414adc89c0102b26ffec9d : SUCCESS in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d700ff19af6b2f21ca3a00a859c2b072c5809bae","revisions":{"d700ff19af6b2f21ca3a00a859c2b072c5809bae":{"kind":"REWORK","_number":1,"created":"2023-08-17 13:30:19.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/41/891741/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/41/891741/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/41/891741/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/41/891741/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/41/891741/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/41/891741/1"}}},"commit":{"parents":[{"commit":"e1a0f2bf963c519378e5e0f6d2151c5b01da8398","subject":"Merge \"Update TOX_CONSTRAINTS_FILE for stable/2023.1\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/e1a0f2bf963c519378e5e0f6d2151c5b01da8398"}]}],"author":{"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","date":"2023-08-02 13:38:42.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2023-08-17 13:30:19.000000000","tz":0},"subject":"Fix lack of constraints for ironic-inspector","message":"Fix lack of constraints for ironic-inspector\n\nOSA playbooks only call this role once for all Ironic containers\n(API and inspector). As a result, the wheel builds only happen\nonce. If the first host (which is responsible for wheel builds)\nis an API container, these vars would prevent Ironic inspector\nrequirements being accounted for, and as such no matching\nconstraints will be generated.\n\nWhen the venv is deployed to the Ironic inspector container,\nthe lack of constraints can cause dependencies which are too new\nto be installed, causing the service to fail.\n\nAlternatively this role could be called twice by the playbook\nfor differing container/host roles, but as inspector is expected\nto be merged into ironic at some point this feels equally valid.\n\nChange-Id: I3952a4e5514824381410d87ed6d535f13ec40498\n(cherry picked from commit a665f45a696514ab01e913374cd921fa88ce6c0d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/d700ff19af6b2f21ca3a00a859c2b072c5809bae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/d700ff19af6b2f21ca3a00a859c2b072c5809bae"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
