)]}'
{"id":"openstack%2Fopenstack-ansible-os_ironic~951003","triplet_id":"openstack%2Fopenstack-ansible-os_ironic~master~I4c570aa4d9d5d6723235cf350c6cb1625def4a44","project":"openstack/openstack-ansible-os_ironic","branch":"master","attention_set":{},"removed_from_attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2025-11-17 09:09:27.000000000","reason":"Change was abandoned"},"25023":{"account":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"last_update":"2025-11-17 09:09:27.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I4c570aa4d9d5d6723235cf350c6cb1625def4a44","subject":"Add apparmor rules for ironic inspector","status":"ABANDONED","created":"2025-05-27 14:25:55.000000000","updated":"2025-11-17 09:09:27.000000000","total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"dabb7bacc8cb64b934cc6403a9b01088eaf22b57","_number":951003,"virtual_id_number":951003,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"all":[{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2025-11-17 09:09:07.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-27 16:28:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-29 14:39:43.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"CC"},{"updated":"2025-06-02 08:47:26.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2025-08-18 21:45:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"}],"messages":[{"id":"0ffc7cc5bde4188462c267bdf371380c5f300e97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-27 14:25:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"99de9cd8ea0b06d28e477fda22b42d76e7e71a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 16:28:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/37b3c8bf968341b295d9e67b0f6c8da1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/53b7a81b4ab246af8ef4e811833aae2d : FAILURE in 8m 29s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/f5f2113977cd4292bc0b402a56b7a8e6 : SUCCESS in 1h 42m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/569df6f3aeef4d598de0d4acc81d15c4 : FAILURE in 1h 30m 52s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/56b1b61d5a7442728de458b4b7f5d679 : FAILURE in 1h 17m 47s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f57b6e9cab72415d86711324b46aa7e7 : FAILURE in 1h 19m 02s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/0a5bc2b2e0ac45cb9e56dfee3292281e : FAILURE in 1h 15m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c57bcc090b44fd496779aef8a291e2c : SUCCESS in 5m 49s","accounts_in_message":[],"_revision_number":1},{"id":"61519b44afb6a0e484e14e8ade3cc8c28940f048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-29 14:38:22.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1cd7b58cb46d4909b4d1649429430b4cc0d8a140","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-05-29 14:39:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2e29b611371fc73265d6997ff7a89285a90acd49","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-05-29 14:51:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c860952a87712d7f902cceb9fa9d6d806cb3bb63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 16:46:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5dcd2aa8a694c72ade201770665c3e6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cfe6f1819b74433098bc20cb6d90e60c : FAILURE in 5m 54s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d27d7295e746404a8a7f1f8a3ec40d6f : SUCCESS in 1h 32m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/f9700aaa09c34e8797182431c3930edf : SUCCESS in 2h 01m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8d12c985f7e34831aafb1e3fdb6b24e3 : SUCCESS in 1h 34m 00s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4a30f29432d44e2974344e292e00c35 : FAILURE in 1h 05m 23s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b4ea034451c144cba30f5324a5f4e2c9 : FAILURE in 1h 14m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1b9d35484174318a6268422774bbf70 : SUCCESS in 5m 41s","accounts_in_message":[],"_revision_number":2},{"id":"12385fef1834fc4a42cfa669aa7c5e3f1d9c522a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-06-02 08:38:14.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4184638efc9c18437656d48f9849720e3ebdbfd0","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-02 08:47:26.000000000","message":"Patch Set 3: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"181c7ba0dfc2ca048c35d931973ed01d042043ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-02 08:51:57.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"6730a759a3d807e2afe01a89b52dca9762928561","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-02 08:52:04.000000000","message":"Patch Set 4: Backport-Candidate+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b239a1e45b4cc5729cbc0b5781685618b80d27b6","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-06-02 08:52:11.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"befcf16de015345cff7348dad618600cf3961c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 11:14:50.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/af744b5c049843d0868665466dec17c0\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e4a2c74aed20469fa71fb55872147891 : SUCCESS in 8m 25s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/689b6947f96549f98a1d7ee3d1e9847d : SUCCESS in 1h 38m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a37b20bdf02f4844bc787c6c137bfde9 : FAILURE in 1h 25m 09s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7389ec19b0dc416fbf145c94e3796508 : FAILURE in 1h 23m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/95489eb11afe4ae69d4474210f8ec522 : FAILURE in 1h 16m 32s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/01293deec4034360b43276692229d62e : FAILURE in 1h 16m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbf4eef0133242068a6a88b5eef26ca7 : SUCCESS in 6m 38s","accounts_in_message":[],"_revision_number":4},{"id":"1a337c2833eec9177b56f0b0f3cd2fcd689d9a12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-06-02 13:45:34.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"8d0abbeb0366e8ca71e527d1433069146ddfb80c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 15:28:09.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/312e0ff1a8b24fcf8782cbaf363cc910\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bcfae7d57e8140fc872b88435f5bf026 : SUCCESS in 8m 16s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/317fea6ebb5d4557811884e9f48e0330 : SUCCESS in 1h 29m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1baec8054b0642768974b2184b58cab6 : SUCCESS in 1h 24m 35s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/97776cd668d24307a8200461a38a9241 : SUCCESS in 1h 39m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7f2122e872db4bff8ec7e98da897d9e3 : FAILURE in 1h 16m 14s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1df2a2e18e474959913e90d48bcd7d84 : FAILURE in 1h 08m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb5e925847e4476ab5a8c91e010ebca8 : SUCCESS in 5m 15s","accounts_in_message":[],"_revision_number":5},{"id":"bd98749346012d3b27ff29c32868e1a557e9de24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2025-07-17 12:35:36.000000000","message":"Patch Set 6: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"d17aa6a13d6f6d2a54f80513e80cc488ae9a0c28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-17 13:22:44.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7cfa18e88d114ab8ae667e441216c992\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/daecea79834d4a0ea5981f730711cc17 : SUCCESS in 4m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/146c5b1d2e25442db205a02b14f14994 : FAILURE in 42m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/bfd76970c95943a1b572cbb030d0fb87 : FAILURE in 41m 03s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/71b915989398484eaa0e4052159886b7 : FAILURE in 44m 19s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7d34dd34793d4b069d3311248f8bf05d : FAILURE in 38m 40s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/df835da285694e3b8655e0e7a99d97c2 : FAILURE in 38m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa887d5b109c415f8e3a33a2d083f50a : SUCCESS in 6m 02s","accounts_in_message":[],"_revision_number":6},{"id":"3a58bc7639078f118fa6b8f2193c4ce16ea881b3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-18 21:45:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"31b25a1e93d852e41ff7a20385edbb1846673c67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-19 00:28:13.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/33e61bcd36a740708578c795528de898\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8fc3a504a6d7482dbec87b2abc644de6 : SUCCESS in 9m 07s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/f938051771c747b6bafb4cec4de31517 : FAILURE in 1h 15m 58s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/a30d1a68d61d4bcc9830ecf178774cb1 : SUCCESS in 1h 41m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/83442655bc2d43d9b9bc143658a87856 : SUCCESS in 1h 53m 08s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/16d39c4f318e43e882b6ca5b29dfb7b7 : FAILURE in 1h 47m 40s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/2ee8ee3e30294f43b458b134b42ffbe7 : SUCCESS in 2h 36m 06s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9e5cf01b739c4e48a63700ebc4577395 : FAILURE in 2h 29m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a1f0517eaa4109bb34e972f8d5ebeb : SUCCESS in 7m 28s","accounts_in_message":[],"_revision_number":6},{"id":"1a9d702e56c99b9e49889f3f4fe0d78b2bee8df3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-10 11:06:22.000000000","message":"Patch Set 7: Commit message was updated.\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"**changekind:NO_CODE_CHANGE** OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"db609cf3dcad3709b629b4e4fc3a91aa923eb9ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-10 13:52:10.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eadb97f87f5e4bd18bf5a737599ac5b6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c3bdb78cd8d0471bbcddf9fc2ebc18fb : SUCCESS in 6m 35s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/3edeaeb3edbf476fb9812c36334a2944 : FAILURE in 48m 40s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/51a1bb68b8374ba6978b58eb84c10a24 : FAILURE in 1h 33m 01s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/f011633509b947208b26347eaff0a546 : FAILURE in 1h 04m 35s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/04300573a8434d26935a36fcaccd9bc7 : FAILURE in 1h 02m 45s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3b4efe6dc5e942478d372a85c70ce56b : FAILURE in 1h 26m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/06603b8dceac4a1289f2bbc08cdd7f33 : FAILURE in 1h 33m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/b85d7391d29948a8b396fadd1d969945 : FAILURE in 1h 37m 08s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ef014cf55bd44627a1bc73e4324f3806 : FAILURE in 2h 38m 25s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/fdbbe9d14c9d4bc79fbe75244d4a4e18 : FAILURE in 2h 07m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d2b6d4e154e47c3a7db73f5b434e2e1 : SUCCESS in 9m 06s","accounts_in_message":[],"_revision_number":7},{"id":"436982c2a0523f99f6051f6cb2e4748a11bd78f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-11 20:19:07.000000000","message":"Patch Set 8: Patch Set 7 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"d1a7d2e953cf98f140cfa9bcadb9fa34bcdbc7de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-11 23:17:33.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/030e0c881eea4799a5e985721e21e2a6\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e1e6da3eb4dd4a0f96aa047bb551b032 : FAILURE in 7m 00s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/e8cd4390f9524b6ca457febac038b523 : SUCCESS in 1h 22m 05s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/131b766a73764221a788fede93c19c8f : SUCCESS in 1h 39m 27s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/53612d6bd5ec48b5aaea62a9030c49bd : SUCCESS in 1h 27m 17s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/43a1cc5132194c16bd995958fe923a39 : SUCCESS in 58m 36s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/1a131b31e2ea493da9f5c8b48c8126a4 : FAILURE in 42m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/daacc1facba54d3483d1a8fee211d7e8 : SUCCESS in 1h 39m 20s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/8bdb37d59f13421197cdc5f23df4885d : SUCCESS in 1h 36m 33s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7b725642782c414f8a8e33cf0e59e829 : SUCCESS in 2h 32m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/a8e5c932a8a045e5ada2ab0a393c2292 : SUCCESS in 2h 35m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64f0d1ba992548358e65ad73e4d20bb0 : SUCCESS in 6m 36s","accounts_in_message":[],"_revision_number":8},{"id":"875eff97180ea60f1d5ab32e55fa6fde9129583e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-12 11:31:44.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"1e45d081d1f308bc7be282556e3a484fcefa2675","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 16:32:50.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14c78221c48747d790e081d43da11cd1\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/6e12720576cc404a80b197498870e6bc : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/c500d8d9f4de400ba037066336afe71f : SUCCESS in 1h 40m 22s\n- openstack-ansible-deploy-aio_metal-debian-bookworm https://zuul.opendev.org/t/openstack/build/d70d645502744eb89cf8c1de15969be9 : SUCCESS in 1h 22m 59s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/d6374d0af7334368ae30a96b838d9b8a : SUCCESS in 1h 30m 47s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/596eb693416e419f9b1a50196b4baca0 : SUCCESS in 55m 26s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/3cf05ddf65c643ca8535632145b188f6 : FAILURE in 1h 27m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/dc2c83788afb4ddbb7b8128834792a7f : FAILURE in 41m 01s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/1189268e2faa4e739fb37811343d8527 : SUCCESS in 55m 29s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/92dc99f1620341829bf97bd9834794e6 : SUCCESS in 2h 35m 53s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cce1abceed054cec9bc7416efc43481e : SUCCESS in 2h 41m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1797b3982c14ef998a49e9683d9707d : SUCCESS in 4m 06s","accounts_in_message":[],"_revision_number":9},{"id":"8894d3b1d31c5778a151c56fef5a3a6e592ebab9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-17 09:09:07.000000000","message":"Patch Set 10: Patch Set 9 was rebased\n\nCopied Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **changekind:TRIVIAL_REBASE** OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"dabb7bacc8cb64b934cc6403a9b01088eaf22b57","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2025-11-17 09:09:27.000000000","message":"Abandoned\n\nsquashed with https://review.opendev.org/c/openstack/openstack-ansible-os_ironic/+/950508/12","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"592583fd77af48c1d1273b71b4a37637fc6e4ae8","revisions":{"d2c7ee52c5ab4ff4134a1a6135672f14010a2455":{"kind":"REWORK","_number":1,"created":"2025-05-27 14:25:55.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/951003/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/03/951003/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/1"}}},"commit":{"parents":[{"commit":"6fb1dba4c95269ff8294d716bb4001fa2f041c35","subject":"Remove tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/6fb1dba4c95269ff8294d716bb4001fa2f041c35"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-27 14:18:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-27 14:18:21.000000000","tz":60},"subject":"Add apparmor rules for ironic inspector","message":"Add apparmor rules for ironic inspector\n\nThis patch adds the minimal amount of apparmor rules required to\nallow the dnsmasq instance for inspector start up. This is necessary\nbecause inspector puts the config file in a non-standard directory\nthat is not covered by the default apparmor rules.\n\nIn addition, fix a permissions error where dnsmasq is not able to\nread its configuration in /etc/ironic-inspector as it was not\nconfigured to drop priviledges from root to a specific user,\nand instead was running as the \"nobody\" user.\n\nThis patch does not excercise the functionality of inspector so\nit is possible that further apparmor rules are required for runtime\nin addition to those added for startup.\n\nChange-Id: I4c570aa4d9d5d6723235cf350c6cb1625def4a44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/d2c7ee52c5ab4ff4134a1a6135672f14010a2455"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/d2c7ee52c5ab4ff4134a1a6135672f14010a2455"}]},"branch":"refs/heads/master"},"17defc737cb76b69ee4a28ddcf55c7e271c33ffd":{"kind":"REWORK","_number":2,"created":"2025-05-29 14:38:22.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/951003/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/03/951003/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/2"}}},"commit":{"parents":[{"commit":"6fb1dba4c95269ff8294d716bb4001fa2f041c35","subject":"Remove tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/6fb1dba4c95269ff8294d716bb4001fa2f041c35"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-27 14:18:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-29 14:38:06.000000000","tz":0},"subject":"Add apparmor rules for ironic inspector","message":"Add apparmor rules for ironic inspector\n\nThis patch adds the minimal amount of apparmor rules required to\nallow the dnsmasq instance for inspector start up. This is necessary\nbecause inspector puts the config file in a non-standard directory\nthat is not covered by the default apparmor rules.\n\nIn addition, fix a permissions error where dnsmasq is not able to\nread its configuration in /etc/ironic-inspector as it was not\nconfigured to drop priviledges from root to a specific user,\nand instead was running as the \"nobody\" user.\n\nThis patch does not excercise the functionality of inspector so\nit is possible that further apparmor rules are required for runtime\nin addition to those added for startup.\n\nChange-Id: I4c570aa4d9d5d6723235cf350c6cb1625def4a44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/17defc737cb76b69ee4a28ddcf55c7e271c33ffd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/17defc737cb76b69ee4a28ddcf55c7e271c33ffd"}]},"branch":"refs/heads/master"},"543fa760ea786f6e77bd440833c7711bb5d8c97d":{"kind":"REWORK","_number":3,"created":"2025-06-02 08:38:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/951003/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/03/951003/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/3"}}},"commit":{"parents":[{"commit":"6fb1dba4c95269ff8294d716bb4001fa2f041c35","subject":"Remove tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/6fb1dba4c95269ff8294d716bb4001fa2f041c35"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-27 14:18:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-06-02 08:38:13.000000000","tz":0},"subject":"Add apparmor rules for ironic inspector","message":"Add apparmor rules for ironic inspector\n\nThis patch adds the minimal amount of apparmor rules required to\nallow the dnsmasq instance for inspector start up. This is necessary\nbecause inspector puts the config file in a non-standard directory\nthat is not covered by the default apparmor rules.\n\nIn addition, fix a permissions error where dnsmasq is not able to\nread its configuration in /etc/ironic-inspector as it was not\nconfigured to drop priviledges from root to a specific user,\nand instead was running as the \"nobody\" user.\n\nThis patch does not excercise the functionality of inspector so\nit is possible that further apparmor rules are required for runtime\nin addition to those added for startup.\n\nChange-Id: I4c570aa4d9d5d6723235cf350c6cb1625def4a44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/543fa760ea786f6e77bd440833c7711bb5d8c97d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/543fa760ea786f6e77bd440833c7711bb5d8c97d"}]},"branch":"refs/heads/master"},"ae4e294af91e4b948c6bc13eb87c622915475882":{"kind":"REWORK","_number":4,"created":"2025-06-02 08:51:57.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/951003/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/03/951003/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/4"}}},"commit":{"parents":[{"commit":"6fb1dba4c95269ff8294d716bb4001fa2f041c35","subject":"Remove tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/6fb1dba4c95269ff8294d716bb4001fa2f041c35"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-27 14:18:21.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-06-02 08:51:57.000000000","tz":0},"subject":"Add apparmor rules for ironic inspector","message":"Add apparmor rules for ironic inspector\n\nThis patch adds the minimal amount of apparmor rules required to\nallow the dnsmasq instance for inspector start up. This is necessary\nbecause inspector puts the config file in a non-standard directory\nthat is not covered by the default apparmor rules.\n\nIn addition, fix a permissions error where dnsmasq is not able to\nread its configuration in /etc/ironic-inspector as it was not\nconfigured to drop priviledges from root to a specific user,\nand instead was running as the \"nobody\" user.\n\nThis patch does not excercise the functionality of inspector so\nit is possible that further apparmor rules are required for runtime\nin addition to those added for startup.\n\nChange-Id: I4c570aa4d9d5d6723235cf350c6cb1625def4a44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/ae4e294af91e4b948c6bc13eb87c622915475882"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/ae4e294af91e4b948c6bc13eb87c622915475882"}]},"branch":"refs/heads/master"},"bd81448d2b822e0d1deae3982ce92fcbceec8421":{"kind":"REWORK","_number":5,"created":"2025-06-02 13:45:34.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/951003/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/03/951003/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/5"}}},"commit":{"parents":[{"commit":"6fb1dba4c95269ff8294d716bb4001fa2f041c35","subject":"Remove tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/6fb1dba4c95269ff8294d716bb4001fa2f041c35"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-27 14:18:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-06-02 13:45:34.000000000","tz":0},"subject":"Add apparmor rules for ironic inspector","message":"Add apparmor rules for ironic inspector\n\nThis patch adds the minimal amount of apparmor rules required to\nallow the dnsmasq instance for inspector start up. This is necessary\nbecause inspector puts the config file in a non-standard directory\nthat is not covered by the default apparmor rules.\n\nIn addition, fix a permissions error where dnsmasq is not able to\nread its configuration in /etc/ironic-inspector as it was not\nconfigured to drop priviledges from root to a specific user,\nand instead was running as the \"nobody\" user.\n\nThis patch does not excercise the functionality of inspector so\nit is possible that further apparmor rules are required for runtime\nin addition to those added for startup.\n\nChange-Id: I4c570aa4d9d5d6723235cf350c6cb1625def4a44\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/bd81448d2b822e0d1deae3982ce92fcbceec8421"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/bd81448d2b822e0d1deae3982ce92fcbceec8421"}]},"branch":"refs/heads/master"},"30feae0c0eb4a440ff7c32d6c4bdde3207c6d407":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-07-17 12:35:36.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/951003/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/03/951003/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/6"}}},"commit":{"parents":[{"commit":"6fb1dba4c95269ff8294d716bb4001fa2f041c35","subject":"Remove tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/6fb1dba4c95269ff8294d716bb4001fa2f041c35"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-27 14:18:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-07-17 12:35:36.000000000","tz":0},"subject":"Add apparmor rules for ironic inspector","message":"Add apparmor rules for ironic inspector\n\nThis patch adds the minimal amount of apparmor rules required to\nallow the dnsmasq instance for inspector start up. This is necessary\nbecause inspector puts the config file in a non-standard directory\nthat is not covered by the default apparmor rules.\n\nIn addition, fix a permissions error where dnsmasq is not able to\nread its configuration in /etc/ironic-inspector as it was not\nconfigured to drop priviledges from root to a specific user,\nand instead was running as the \"nobody\" user.\n\nThis patch does not excercise the functionality of inspector so\nit is possible that further apparmor rules are required for runtime\nin addition to those added for startup.\n\nChange-Id: I4c570aa4d9d5d6723235cf350c6cb1625def4a44\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/955226\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/30feae0c0eb4a440ff7c32d6c4bdde3207c6d407"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/30feae0c0eb4a440ff7c32d6c4bdde3207c6d407"}]},"branch":"refs/heads/master","description":"Edit commit message"},"2f37c4a03b256af016ba9b538c12f3bdcf40cf6e":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-11-10 11:06:22.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/951003/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/03/951003/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/7"}}},"commit":{"parents":[{"commit":"6fb1dba4c95269ff8294d716bb4001fa2f041c35","subject":"Remove tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/6fb1dba4c95269ff8294d716bb4001fa2f041c35"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-27 14:18:21.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-11-10 11:06:22.000000000","tz":0},"subject":"Add apparmor rules for ironic inspector","message":"Add apparmor rules for ironic inspector\n\nThis patch adds the minimal amount of apparmor rules required to\nallow the dnsmasq instance for inspector start up. This is necessary\nbecause inspector puts the config file in a non-standard directory\nthat is not covered by the default apparmor rules.\n\nIn addition, fix a permissions error where dnsmasq is not able to\nread its configuration in /etc/ironic-inspector as it was not\nconfigured to drop priviledges from root to a specific user,\nand instead was running as the \"nobody\" user.\n\nThis patch does not excercise the functionality of inspector so\nit is possible that further apparmor rules are required for runtime\nin addition to those added for startup.\n\nChange-Id: I4c570aa4d9d5d6723235cf350c6cb1625def4a44\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/966515\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/2f37c4a03b256af016ba9b538c12f3bdcf40cf6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/2f37c4a03b256af016ba9b538c12f3bdcf40cf6e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"312810ea51c7e6d73b0f8f5fc4986fbb8730a170":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-11-11 20:19:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/951003/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/03/951003/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/8"}}},"commit":{"parents":[{"commit":"46fcab0391cb87b7c46bd42fadc13a2996dcb1e6","subject":"Switch from wsgi script to wsgi module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/46fcab0391cb87b7c46bd42fadc13a2996dcb1e6"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-27 14:18:21.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-11-11 20:19:07.000000000","tz":0},"subject":"Add apparmor rules for ironic inspector","message":"Add apparmor rules for ironic inspector\n\nThis patch adds the minimal amount of apparmor rules required to\nallow the dnsmasq instance for inspector start up. This is necessary\nbecause inspector puts the config file in a non-standard directory\nthat is not covered by the default apparmor rules.\n\nIn addition, fix a permissions error where dnsmasq is not able to\nread its configuration in /etc/ironic-inspector as it was not\nconfigured to drop priviledges from root to a specific user,\nand instead was running as the \"nobody\" user.\n\nThis patch does not excercise the functionality of inspector so\nit is possible that further apparmor rules are required for runtime\nin addition to those added for startup.\n\nChange-Id: I4c570aa4d9d5d6723235cf350c6cb1625def4a44\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/966515\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/312810ea51c7e6d73b0f8f5fc4986fbb8730a170"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/312810ea51c7e6d73b0f8f5fc4986fbb8730a170"}]},"branch":"refs/heads/master","description":"Rebase"},"7fbf247bc0cc4a65845cde344c18373bd3a4fd12":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-11-12 11:31:44.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/951003/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/03/951003/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/9"}}},"commit":{"parents":[{"commit":"28e815ffb1f1ac1a882b6ad9d461751820663aec","subject":"Switch from wsgi script to wsgi module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/28e815ffb1f1ac1a882b6ad9d461751820663aec"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-27 14:18:21.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-11-12 11:31:44.000000000","tz":0},"subject":"Add apparmor rules for ironic inspector","message":"Add apparmor rules for ironic inspector\n\nThis patch adds the minimal amount of apparmor rules required to\nallow the dnsmasq instance for inspector start up. This is necessary\nbecause inspector puts the config file in a non-standard directory\nthat is not covered by the default apparmor rules.\n\nIn addition, fix a permissions error where dnsmasq is not able to\nread its configuration in /etc/ironic-inspector as it was not\nconfigured to drop priviledges from root to a specific user,\nand instead was running as the \"nobody\" user.\n\nThis patch does not excercise the functionality of inspector so\nit is possible that further apparmor rules are required for runtime\nin addition to those added for startup.\n\nChange-Id: I4c570aa4d9d5d6723235cf350c6cb1625def4a44\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/966515\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/7fbf247bc0cc4a65845cde344c18373bd3a4fd12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/7fbf247bc0cc4a65845cde344c18373bd3a4fd12"}]},"branch":"refs/heads/master","description":"Rebase"},"592583fd77af48c1d1273b71b4a37637fc6e4ae8":{"kind":"TRIVIAL_REBASE","_number":10,"created":"2025-11-17 09:09:07.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/03/951003/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_ironic","ref":"refs/changes/03/951003/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_ironic refs/changes/03/951003/10"}}},"commit":{"parents":[{"commit":"95938b77d1af7cdbb767b5d822652c62268c2b5d","subject":"Switch from wsgi script to wsgi module and add apparmor rules","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/95938b77d1af7cdbb767b5d822652c62268c2b5d"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2025-05-27 14:18:21.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","date":"2025-11-17 09:09:07.000000000","tz":0},"subject":"Add apparmor rules for ironic inspector","message":"Add apparmor rules for ironic inspector\n\nThis patch adds the minimal amount of apparmor rules required to\nallow the dnsmasq instance for inspector start up. This is necessary\nbecause inspector puts the config file in a non-standard directory\nthat is not covered by the default apparmor rules.\n\nIn addition, fix a permissions error where dnsmasq is not able to\nread its configuration in /etc/ironic-inspector as it was not\nconfigured to drop priviledges from root to a specific user,\nand instead was running as the \"nobody\" user.\n\nThis patch does not excercise the functionality of inspector so\nit is possible that further apparmor rules are required for runtime\nin addition to those added for startup.\n\nChange-Id: I4c570aa4d9d5d6723235cf350c6cb1625def4a44\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible-openstack_hosts/+/966515\nSigned-off-by: Jonathan Rosser \u003cjonathan.rosser@rd.bbc.co.uk\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/592583fd77af48c1d1273b71b4a37637fc6e4ae8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_ironic/commit/592583fd77af48c1d1273b71b4a37637fc6e4ae8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
