)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~1003542","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ie66c3b18b8354f4f1e80b707c813a8f700c46f20","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"osa/galera_readonly","attention_set":{"28619":{"account":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"last_update":"2026-09-18 09:27:28.000000000","reason":"\u003cGERRIT_ACCOUNT_31542\u003e replied on the change","reason_account":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie66c3b18b8354f4f1e80b707c813a8f700c46f20","subject":"Implement readonly connection for MariaDB","status":"NEW","created":"2026-09-02 15:01:31.000000000","updated":"2026-09-18 17:03:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a020d7ea3bcefa530c8af534fd41ac36183634a5","_number":1003542,"virtual_id_number":1003542,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-03 15:42:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2026-09-18 09:27:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2026-09-18 17:03:44.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-09-02 20:25:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-09-18 09:27:28.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2026-09-18 17:03:44.000000000","updated_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"reviewer":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"state":"REVIEWER"}],"messages":[{"id":"47cdd4a22c37095b6165101b83b1eee39fa5b5c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-02 15:01:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2c4cb2a28ca8d41f7c4dadd92782f7ffde4dffab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-02 15:35:26.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"89e8b847b554135bb7559f9ec662288c74f65e03","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-02 17:12:04.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"122e500924448f3d1097bfd8c02a6a483a7f8986","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-02 20:25:32.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92110a26f2b843dc86e772c0a3dbd593\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7109162cfaa54c8086a7c786d276aec6 : SUCCESS in 13m 52s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/d794916954704d5cba21a660cab4daa1 : SUCCESS in 1h 15m 32s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/0308e10a3b1645ecb37a2b6ced9c6d3e : SUCCESS in 1h 24m 29s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/56f7040b47244be09004ba65e75a0d93 : SUCCESS in 56m 10s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e51098fa3e354104b2e26e551c4713c5 : SUCCESS in 59m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/9c559ac8d5144005a3da722d820009cf : SUCCESS in 1h 16m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/4491de2e2cb64c68a6586240e8477e6e : SUCCESS in 1h 27m 43s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/77b0d09c71dc44ea8711ab00287d4ca5 : SUCCESS in 1h 34m 28s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/fa36f187c0a349eab803242b865a2c3c : SUCCESS in 2h 00m 58s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/088b64a148a3430ab4ce48933f92e047 : FAILURE in 14m 01s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/ada56874606f44b38311a6f48f6546b8 : SUCCESS in 2h 24m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/703f664e0f5a42ed939b58457d62e6eb : SUCCESS in 51m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/b0597676328d4e609de931d6325df484 : SUCCESS in 1h 30m 15s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/9d401111c8264659977d6e7d1c6c4c6b : SUCCESS in 55m 57s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/82cfe5c7f5c7423e98f92ecc4634d50e : SUCCESS in 1h 36m 22s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c71c94e77e8a4493b7e2d1b9eccc68f9 : SUCCESS in 1h 04m 25s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/e5158b7a365c4e28896368386f261449 : SUCCESS in 1h 05m 05s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7296028a97af430ea813ba665fca3fc8 : SUCCESS in 1h 34m 07s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/7408638aa7a74a35ba24204dcdc55b61 : SUCCESS in 1h 25m 58s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/33cf7674486a45f3a12a5c159ff291ce : SUCCESS in 1h 17m 37s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5bcc02550ac149f6b849c8f26e0a2bba : SUCCESS in 1h 35m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01e472034fba4af7b95354fce583c619 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab528f5f5f2a48fd845b3e838f27f255 : SUCCESS in 3m 53s","accounts_in_message":[],"_revision_number":3},{"id":"2193e55ee31f109137275f1137fb27663d325cdb","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-03 09:30:17.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6570231ad55f36bb56b413f90e6cac37d3a4d815","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2026-09-03 13:38:45.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a48197d0e1dff63991949be616a3f525cc7c733c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-03 15:42:28.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/504adf05c1a7482db9ccfbb2e259b745\n\n- openstack-ansible-linters-ubuntu-noble https://zuul.opendev.org/t/openstack/build/618983b5011f458d8ce9fa79a9aaee41 : SUCCESS in 15m 04s\n- openstack-ansible-deploy-aio_metal-centos-10-stream https://zuul.opendev.org/t/openstack/build/0615cc82e3494a7f8f172051d4486e09 : SUCCESS in 1h 29m 18s\n- openstack-ansible-deploy-aio_metal-debian-trixie https://zuul.opendev.org/t/openstack/build/668134e920354c3c994afdd21f0dddd3 : SUCCESS in 1h 25m 20s\n- openstack-ansible-deploy-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/640ea2e3fa694e68a6a5fa58a57c26a3 : SUCCESS in 1h 31m 40s\n- openstack-ansible-deploy-aio_metal_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/9fb1da85699048478bf79d34cb825d4a : SUCCESS in 1h 37m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/371adde173374f9993f6edfd16cff561 : SUCCESS in 1h 30m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/77128d80c0ac4f9a91ec0f35288dd467 : SUCCESS in 1h 24m 29s\n- openstack-ansible-deploy-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/eba9256e049543589099d4dc4c258d4c : SUCCESS in 1h 28m 56s\n- openstack-ansible-upgrade-aio_metal-rockylinux-10 https://zuul.opendev.org/t/openstack/build/c7959c93775a48d491714608c5d80eb7 : SUCCESS in 2h 17m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/5ebb22ab53ee4bdc91b22a247389a203 : SUCCESS in 2h 32m 12s\n- openstack-ansible-upgrade-aio_metal_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/7df3457a681e410e96dfb86c7f0f0e00 : SUCCESS in 2h 19m 47s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-noble https://zuul.opendev.org/t/openstack/build/52e5a7387e0f43b583d7a72129e95d52 : SUCCESS in 1h 22m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/650c1c5449a04815ad16f00258846954 : SUCCESS in 1h 24m 33s\n- openstack-ansible-deploy-infra_lxc-centos-10-stream https://zuul.opendev.org/t/openstack/build/153a523e53fa4855a6f7cbc4466b613b : SUCCESS in 1h 45m 17s\n- openstack-ansible-deploy-infra_lxc-debian-trixie https://zuul.opendev.org/t/openstack/build/6a72cb2487574afc9db3c42323050952 : SUCCESS in 1h 41m 36s\n- openstack-ansible-deploy-infra_lxc-rockylinux-10 https://zuul.opendev.org/t/openstack/build/0aa8e7646a6c48ef9bf7ef3b8c93215c : SUCCESS in 1h 51m 40s\n- openstack-ansible-deploy-infra_lxc_tls-rockylinux-10 https://zuul.opendev.org/t/openstack/build/8d3f2bfeb45c4e06ae9538df6f90da08 : SUCCESS in 1h 53m 54s\n- openstack-ansible-deploy-infra_lxc-ubuntu-noble https://zuul.opendev.org/t/openstack/build/e36c5c8ef129401593110f1fcb0cbd1f : SUCCESS in 1h 37m 44s\n- openstack-ansible-deploy-infra_lxc-ubuntu-resolute https://zuul.opendev.org/t/openstack/build/6749961a801d4d2e9338a3348f892cc7 : SUCCESS in 1h 41m 52s\n- openstack-ansible-deploy-infra_lxc_stepca-ubuntu-noble https://zuul.opendev.org/t/openstack/build/cd68f42c90454b4ba67ea54c285c5d7c : SUCCESS in 1h 36m 08s\n- openstack-ansible-deploy-infra_lxc_tls-ubuntu-noble https://zuul.opendev.org/t/openstack/build/c889004a5378441a8704222379532edb : SUCCESS in 1h 44m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b7e5da2bf2e459aad085a013bf002c5 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a7bfc30d9ea457ca22c1b5496212972 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":3},{"id":"04f8a5e05fc4671ac6169bdd3dac27cdb17606e4","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2026-09-18 09:27:28.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"a020d7ea3bcefa530c8af534fd41ac36183634a5","author":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"},"date":"2026-09-18 17:03:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dd62531071894bed1f0a6646c99a4098d0f29271","revisions":{"15418c5d46219f56cfbc84dcbe70d4b12770fcc7":{"kind":"REWORK","_number":1,"created":"2026-09-02 15:01:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/1003542/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/42/1003542/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/42/1003542/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/42/1003542/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/42/1003542/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/42/1003542/1"}}},"commit":{"parents":[{"commit":"463226f0893bf3767ffdae3532223c440ef8d22c","subject":"Merge \"Update ansible_managed variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/463226f0893bf3767ffdae3532223c440ef8d22c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 15:01:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 15:01:19.000000000","tz":120},"subject":"Implement readonly connection for MariaDB","message":"Implement readonly connection for MariaDB\n\nWhen MariaDB is run in a multu-node Galera cluster we add\nslave_connection to the [database] section.\nThis ensures that reads are pointed towards alternative port/frontend on\nHAProxy.\n\nsynchronous_reader is disabled whenever readonly connection is activated\nas otherwise synchronous reads will go through the connection and not\nslave_connection.\nWith that we are enabling wsrep_sync_wait to ensure consistency on\nMariaDB.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/1003535\nChange-Id: Ie66c3b18b8354f4f1e80b707c813a8f700c46f20\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/15418c5d46219f56cfbc84dcbe70d4b12770fcc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/15418c5d46219f56cfbc84dcbe70d4b12770fcc7"}]},"branch":"refs/heads/master"},"0d488e3a1d28a91773ac8b2860cea68775793c52":{"kind":"REWORK","_number":2,"created":"2026-09-02 15:35:26.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/1003542/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/42/1003542/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/42/1003542/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/42/1003542/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/42/1003542/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/42/1003542/2"}}},"commit":{"parents":[{"commit":"463226f0893bf3767ffdae3532223c440ef8d22c","subject":"Merge \"Update ansible_managed variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/463226f0893bf3767ffdae3532223c440ef8d22c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 15:01:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 15:35:26.000000000","tz":0},"subject":"Implement readonly connection for MariaDB","message":"Implement readonly connection for MariaDB\n\nWhen MariaDB is run in a multu-node Galera cluster we add\nslave_connection to the [database] section.\nThis ensures that reads are pointed towards alternative port/frontend on\nHAProxy.\n\nsynchronous_reader is disabled whenever readonly connection is activated\nas otherwise synchronous reads will go through the connection and not\nslave_connection.\nWith that we are enabling wsrep_sync_wait to ensure consistency on\nMariaDB.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/1003535\nChange-Id: Ie66c3b18b8354f4f1e80b707c813a8f700c46f20\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0d488e3a1d28a91773ac8b2860cea68775793c52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0d488e3a1d28a91773ac8b2860cea68775793c52"}]},"branch":"refs/heads/master"},"dd62531071894bed1f0a6646c99a4098d0f29271":{"kind":"REWORK","_number":3,"created":"2026-09-02 17:12:04.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/42/1003542/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/42/1003542/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/42/1003542/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/42/1003542/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/42/1003542/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/42/1003542/3"}}},"commit":{"parents":[{"commit":"463226f0893bf3767ffdae3532223c440ef8d22c","subject":"Merge \"Update ansible_managed variable\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/463226f0893bf3767ffdae3532223c440ef8d22c"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 15:01:19.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@cleura.com","date":"2026-09-02 17:12:04.000000000","tz":0},"subject":"Implement readonly connection for MariaDB","message":"Implement readonly connection for MariaDB\n\nWhen MariaDB is run in a multu-node Galera cluster we add\nslave_connection to the [database] section.\nThis ensures that reads are pointed towards alternative port/frontend on\nHAProxy.\n\nsynchronous_reader is disabled whenever readonly connection is activated\nas otherwise synchronous reads will go through the connection and not\nslave_connection.\nWith that we are enabling wsrep_sync_wait to ensure consistency on\nMariaDB.\n\nDepends-On: https://review.opendev.org/c/openstack/openstack-ansible/+/1003535\nChange-Id: Ie66c3b18b8354f4f1e80b707c813a8f700c46f20\nSigned-off-by: Dmitriy Rabotyagov \u003cdmitriy.rabotyagov@cleura.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/dd62531071894bed1f0a6646c99a4098d0f29271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/dd62531071894bed1f0a6646c99a4098d0f29271"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":32666,"name":"Damian Dąbrowski","email":"damian@dabrowski.cloud","username":"ddabrowski"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
