)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~620324","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I6fb099cee10ef76f2cd6d20a03ffe53f45a1f85d","project":"openstack/openstack-ansible-os_keystone","branch":"master","hashtags":[],"change_id":"I6fb099cee10ef76f2cd6d20a03ffe53f45a1f85d","subject":"Deployment host can provide shibboleth SP key-pair","status":"MERGED","created":"2018-11-27 14:25:06.000000000","updated":"2019-10-07 19:24:10.000000000","submitted":"2019-10-07 19:22:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"620324-1570476146663-fe29816b","meta_rev_id":"5743b4b40959fb28fd40ceef827c22be0041ce03","_number":620324,"virtual_id_number":620324,"owner":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29584,"name":"Gaultier HUBERT","email":"gaultier@clear-cloud-services.fr"},{"value":0,"_account_id":29589,"name":"Laurent SANZ","email":"laurent.sanz@gmail.com"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":29594,"name":"Taktshang","email":"alexis.tiger@laposte.net"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-07 19:22:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28590,"name":"Mathieu","email":"gaubert.mathieu@gmail.com"},{"value":0,"date":"2019-09-23 09:54:14.000000000","_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":29584,"name":"Gaultier HUBERT","email":"gaultier@clear-cloud-services.fr"},{"value":0,"_account_id":29589,"name":"Laurent SANZ","email":"laurent.sanz@gmail.com"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":29594,"name":"Taktshang","email":"alexis.tiger@laposte.net"},{"value":2,"date":"2019-09-24 09:57:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28590,"name":"Mathieu","email":"gaubert.mathieu@gmail.com"},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":2,"date":"2019-10-07 17:38:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":29584,"name":"Gaultier HUBERT","email":"gaultier@clear-cloud-services.fr"},{"value":0,"_account_id":29589,"name":"Laurent SANZ","email":"laurent.sanz@gmail.com"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":29594,"name":"Taktshang","email":"alexis.tiger@laposte.net"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28590,"name":"Mathieu","email":"gaubert.mathieu@gmail.com"},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":1,"date":"2019-10-07 17:38:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29584,"name":"Gaultier HUBERT","email":"gaultier@clear-cloud-services.fr"},{"value":0,"_account_id":29589,"name":"Laurent SANZ","email":"laurent.sanz@gmail.com"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":29594,"name":"Taktshang","email":"alexis.tiger@laposte.net"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28590,"name":"Mathieu","email":"gaubert.mathieu@gmail.com"},{"value":0,"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28590,"name":"Mathieu","email":"gaubert.mathieu@gmail.com"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29584,"name":"Gaultier HUBERT","email":"gaultier@clear-cloud-services.fr"},{"_account_id":29589,"name":"Laurent SANZ","email":"laurent.sanz@gmail.com"},{"_account_id":29594,"name":"Taktshang","email":"alexis.tiger@laposte.net"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-11-27 15:26:14.000000000","updated_by":{"_account_id":29584,"name":"Gaultier HUBERT","email":"gaultier@clear-cloud-services.fr"},"reviewer":{"_account_id":29584,"name":"Gaultier HUBERT","email":"gaultier@clear-cloud-services.fr"},"state":"REVIEWER"},{"updated":"2018-11-27 16:30:01.000000000","updated_by":{"_account_id":28590,"name":"Mathieu","email":"gaubert.mathieu@gmail.com"},"reviewer":{"_account_id":28590,"name":"Mathieu","email":"gaubert.mathieu@gmail.com"},"state":"REVIEWER"},{"updated":"2018-11-28 09:16:38.000000000","updated_by":{"_account_id":29589,"name":"Laurent SANZ","email":"laurent.sanz@gmail.com"},"reviewer":{"_account_id":29589,"name":"Laurent SANZ","email":"laurent.sanz@gmail.com"},"state":"REVIEWER"},{"updated":"2018-11-28 17:10:05.000000000","updated_by":{"_account_id":29594,"name":"Taktshang","email":"alexis.tiger@laposte.net"},"reviewer":{"_account_id":29594,"name":"Taktshang","email":"alexis.tiger@laposte.net"},"state":"REVIEWER"},{"updated":"2019-01-15 10:54:17.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-09-24 09:57:11.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-10-07 17:38:12.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-10-07 19:22:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ab6c1c2b51ea3a75af13e3310cf3edf78c2ce6e","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2018-11-27 14:25:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"175f24a1efa86ddb6211826d18a29aac6cce8d91","author":{"_account_id":29584,"name":"Gaultier HUBERT","email":"gaultier@clear-cloud-services.fr"},"date":"2018-11-27 15:26:14.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ee6cb04c0af7a489b928106152ce0f7d417570e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-11-27 15:59:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/24/620324/1/check/openstack-ansible-linters/aa30000/ : SUCCESS in 4m 52s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/620324/1/check/openstack-ansible-functional-centos-7/b87ba49/ : SUCCESS in 37m 54s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/620324/1/check/openstack-ansible-functional-opensuse-150/13cfd5d/ : SUCCESS in 38m 00s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/620324/1/check/openstack-ansible-functional-opensuse-423/6ce025f/ : SUCCESS in 32m 56s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/620324/1/check/openstack-ansible-functional-ubuntu-bionic/092e6e8/ : SUCCESS in 31m 28s\n- openstack-tox-docs http://logs.openstack.org/24/620324/1/check/openstack-tox-docs/75fe3fb/html/ : SUCCESS in 2m 55s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/620324/1/check/openstack-ansible-functional-distro_install-centos-7/bfb4876/ : SUCCESS in 33m 10s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/620324/1/check/openstack-ansible-functional-distro_install-opensuse-150/4701d80/ : SUCCESS in 33m 03s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/620324/1/check/openstack-ansible-functional-distro_install-opensuse-423/afcfbaf/ : SUCCESS in 30m 11s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/620324/1/check/openstack-ansible-functional-distro_install-ubuntu-bionic/86fe16c/ : FAILURE in 30m 58s (non-voting)\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/24/620324/1/check/openstack-ansible-uw_apache-centos-7/047da59/ : SUCCESS in 38m 02s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/24/620324/1/check/openstack-ansible-uw_apache-ubuntu-xenial/463ab67/ : SUCCESS in 36m 30s","accounts_in_message":[],"_revision_number":1},{"id":"abd58a61a92eb911592def271b77e7a12e37bb6d","author":{"_account_id":28590,"name":"Mathieu","email":"gaubert.mathieu@gmail.com"},"date":"2018-11-27 16:30:01.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"18bdd00c775d981e886b31f1f7f8736947332bb2","author":{"_account_id":29589,"name":"Laurent SANZ","email":"laurent.sanz@gmail.com"},"date":"2018-11-28 09:16:38.000000000","message":"Patch Set 1: Code-Review+1\n\nusefull","accounts_in_message":[],"_revision_number":1},{"id":"d71118303ea0d1e0d93d4d8cf12499d48b42c3ed","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-01-15 10:48:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"54a0d07c53688a01e61cd0b0d5396c33293f2b5b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-15 11:35:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/24/620324/2/check/openstack-ansible-linters/04218e9/ : SUCCESS in 9m 17s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/24/620324/2/check/openstack-ansible-functional-centos-7/87edf5a/ : SUCCESS in 33m 53s\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/24/620324/2/check/openstack-ansible-functional-opensuse-150/a2ff501/ : SUCCESS in 41m 44s (non-voting)\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/24/620324/2/check/openstack-ansible-functional-opensuse-423/241db03/ : SUCCESS in 32m 03s\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/24/620324/2/check/openstack-ansible-functional-ubuntu-bionic/0711aab/ : SUCCESS in 29m 44s\n- openstack-tox-docs http://logs.openstack.org/24/620324/2/check/openstack-tox-docs/b1bf2d0/html/ : SUCCESS in 3m 04s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/24/620324/2/check/openstack-ansible-functional-distro_install-centos-7/46202e9/ : SUCCESS in 36m 55s\n- openstack-ansible-functional-distro_install-opensuse-150 http://logs.openstack.org/24/620324/2/check/openstack-ansible-functional-distro_install-opensuse-150/28a60b6/ : SUCCESS in 34m 08s (non-voting)\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/24/620324/2/check/openstack-ansible-functional-distro_install-opensuse-423/0358dcb/ : SUCCESS in 29m 48s (non-voting)\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/24/620324/2/check/openstack-ansible-functional-distro_install-ubuntu-bionic/63dee5f/ : SUCCESS in 31m 29s (non-voting)\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/24/620324/2/check/openstack-ansible-uw_apache-centos-7/e5e3ddb/ : SUCCESS in 36m 20s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/24/620324/2/check/openstack-ansible-uw_apache-ubuntu-xenial/e4e360a/ : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":2},{"id":"3d713040940a377bd0ec32e17a8d48c15eab94fd","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-07-11 12:42:06.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"608836192d5b448c408158657917e3433fdd1c17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 14:43:12.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/24/620324/3/check/openstack-ansible-linters/b4e894d/ : SUCCESS in 6m 05s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/24/620324/3/check/openstack-ansible-deploy-aio_metal-debian-stable/1e34951/ : SUCCESS in 56m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/24/620324/3/check/openstack-ansible-deploy-aio_metal-centos-7/c1c4979/ : SUCCESS in 1h 09m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/24/620324/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/9d34ef2/ : SUCCESS in 1h 14m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/24/620324/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2d2bc64/ : SUCCESS in 1h 07m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/24/620324/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/b87252c/ : SUCCESS in 1h 05m 49s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/24/620324/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/6531de6/ : FAILURE in 32m 28s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/24/620324/3/check/openstack-tox-docs/07ca8e0/html/ : SUCCESS in 2m 44s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/24/620324/3/check/openstack-ansible-uw_apache-centos-7/8086237/ : SUCCESS in 38m 24s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/24/620324/3/check/openstack-ansible-uw_apache-ubuntu-bionic/1aaa1f7/ : SUCCESS in 33m 07s","accounts_in_message":[],"_revision_number":3},{"id":"6329513125bb431a8997c790f1df02a65934df8b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-20 10:57:58.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nAlso please, add a release note regarding this","accounts_in_message":[],"_revision_number":3},{"id":"203832274f00e8e9fd2dd95a7b94406c7f281a46","author":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"date":"2019-09-23 09:54:14.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4491447ad85de66d2bd104c0071ff9e7b1dffc42","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-09-24 09:57:11.000000000","message":"Patch Set 3: Code-Review+2\n\nok, that may work out.","accounts_in_message":[],"_revision_number":3},{"id":"8396aebba887595f6c050f79893c819349c9d2b0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-10-07 17:38:12.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"e1de581791dab440d8a07c677f9fbda0f4b46e66","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 17:38:24.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bd25e05a8ca1785e607efdb430d475a7d6aa03ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 19:22:26.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3e9cc42ffaf04d6982cefd0c60dd5733 : SUCCESS in 6m 22s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/47c62e97ca444bb5946733560d596cde : SUCCESS in 1h 28m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aa844ac284f4459eb2569bf05e2a5e77 : SUCCESS in 1h 37m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/64e8003b7e4343d487005f4be3571d5c : SUCCESS in 1h 25m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/018376c2c0c04a92ad239320f6ecef65 : SUCCESS in 1h 09m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3d11d847e062447f9eadcf715b7c1425 : SUCCESS in 1h 07m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f68ee7fd76864ef0b0a141c07c731928 : SUCCESS in 51m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b51ae61de73429a99ef819ea6a0b079 : SUCCESS in 3m 59s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/a477b501d1a242a69e6ea7f9154d0f2c : SUCCESS in 39m 16s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/387eb0e1cbd348cea3aa21672a983366 : SUCCESS in 28m 22s","accounts_in_message":[],"_revision_number":3},{"id":"0ac1b850784424a54437e7a294b8152f03b85ad5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 19:22:26.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"5743b4b40959fb28fd40ceef827c22be0041ce03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-07 19:24:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0afe834f2874d8fae8e081dcfd98a37 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b035c3d0fc506a9a1bf5c475b05c6bf558a41bb1","revisions":{"2f5ba847e0139ed984d54904b80127a248299070":{"kind":"REWORK","_number":1,"created":"2018-11-27 14:25:06.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/24/620324/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/24/620324/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/620324/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/620324/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/620324/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/620324/1"}}},"commit":{"parents":[{"commit":"98f6740218acb08f9a13588a190c6336cdb10a47","subject":"Merge \"Do not remove SSO callback template file\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/98f6740218acb08f9a13588a190c6336cdb10a47"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@osones.com","date":"2018-08-02 13:07:01.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@osones.com","date":"2018-11-27 14:24:38.000000000","tz":60},"subject":"Deployment host can provide shibboleth SP key-pair","message":"Deployment host can provide shibboleth SP key-pair\n\nIf you don\u0027t want to generate shibboleth SP key-pair on first playbook run\nYou can provide sp-cert.pem and sp-key.pem in /etc/openstack_deploy/keystone/\n\nChange-Id: I6fb099cee10ef76f2cd6d20a03ffe53f45a1f85d\nSigned-off-by: Kourosh Vivan \u003ckourosh.vivan@osones.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2f5ba847e0139ed984d54904b80127a248299070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2f5ba847e0139ed984d54904b80127a248299070"}]},"branch":"refs/heads/master"},"2712890c157ba83e1f29df5ea99c73db975495ef":{"kind":"REWORK","_number":2,"created":"2019-01-15 10:48:17.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/24/620324/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/24/620324/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/620324/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/620324/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/620324/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/620324/2"}}},"commit":{"parents":[{"commit":"d40f3bb4124f0ca44f0afc85dbc5322b3c7b276d","subject":"Merge \"ldap: include tasks only when configuration is provided\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d40f3bb4124f0ca44f0afc85dbc5322b3c7b276d"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@osones.com","date":"2018-08-02 13:07:01.000000000","tz":120},"committer":{"name":"Kourosh Vivan","email":"kourosh.vivan@osones.com","date":"2019-01-15 10:47:08.000000000","tz":60},"subject":"Deployment host can provide shibboleth SP key-pair","message":"Deployment host can provide shibboleth SP key-pair\n\nIf you don\u0027t want to generate shibboleth SP key-pair on first playbook run\nYou can provide sp-cert.pem and sp-key.pem in /etc/openstack_deploy/keystone/\n\nChange-Id: I6fb099cee10ef76f2cd6d20a03ffe53f45a1f85d\nSigned-off-by: Kourosh Vivan \u003ckourosh.vivan@osones.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2712890c157ba83e1f29df5ea99c73db975495ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2712890c157ba83e1f29df5ea99c73db975495ef"}]},"branch":"refs/heads/master"},"b035c3d0fc506a9a1bf5c475b05c6bf558a41bb1":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-07-11 12:42:06.000000000","uploader":{"_account_id":23182,"name":"kourosh vivan","email":"kourosh.vivan@ovhcloud.com","username":"KouroshVivan"},"ref":"refs/changes/24/620324/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/24/620324/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/620324/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/620324/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/620324/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/620324/3"}}},"commit":{"parents":[{"commit":"5132d6a83999fe5f76f62a06ce0d296f427125ab","subject":"Merge \"Update distro package names for openSUSE Leap 15.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5132d6a83999fe5f76f62a06ce0d296f427125ab"}]}],"author":{"name":"Kourosh Vivan","email":"kourosh.vivan@osones.com","date":"2018-08-02 13:07:01.000000000","tz":120},"committer":{"name":"kourosh vivan","email":"kourosh.vivan@alterway.fr","date":"2019-07-11 12:42:06.000000000","tz":0},"subject":"Deployment host can provide shibboleth SP key-pair","message":"Deployment host can provide shibboleth SP key-pair\n\nIf you don\u0027t want to generate shibboleth SP key-pair on first playbook run\nYou can provide sp-cert.pem and sp-key.pem in /etc/openstack_deploy/keystone/\n\nChange-Id: I6fb099cee10ef76f2cd6d20a03ffe53f45a1f85d\nSigned-off-by: Kourosh Vivan \u003ckourosh.vivan@osones.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b035c3d0fc506a9a1bf5c475b05c6bf558a41bb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b035c3d0fc506a9a1bf5c475b05c6bf558a41bb1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
