)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~653985","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I6a6660bd3e15908adad9cd1fd76f582506562a97","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"osa/opendev","hashtags":[],"change_id":"I6a6660bd3e15908adad9cd1fd76f582506562a97","subject":"Replace git.openstack.org with opendev.org","status":"MERGED","created":"2019-04-20 12:19:24.000000000","updated":"2019-05-24 20:11:03.000000000","submitted":"2019-05-24 20:11:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"653985-1558728663784-66e7a990","meta_rev_id":"32666784a75b19b9666c1b68bfe926f27e1dc7cc","_number":653985,"virtual_id_number":653985,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-24 20:11:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"recommended":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"all":[{"value":1,"date":"2019-05-24 17:12:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":1,"date":"2019-05-24 17:12:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":2,"date":"2019-05-24 17:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-24 17:12:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":1,"date":"2019-05-24 17:30:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"value":0,"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-20 14:05:21.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-12 16:03:17.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-05-20 20:21:08.000000000","updated_by":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"reviewer":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"state":"REVIEWER"},{"updated":"2019-05-20 20:37:42.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"REVIEWER"},{"updated":"2019-05-20 21:10:44.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-24 20:11:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ebad18a5b870c9537b7f24b68a7b219f2dd4bbb8","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-20 12:19:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19bb5ca3562414ebaa5b9453842a53d484a6fa3a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-20 12:45:13.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"7a9d4a247352d7be9edd8a8b675a659295c4438f","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-20 13:07:31.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"d1a197867b52813f9fe031db93f34cff39173616","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 13:08:19.000000000","message":"Patch Set 2: Verified-1\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"0bc235f3f28b18c960b791fafc47d9321b36604c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-20 13:14:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c680cfc5119dac4fc29e9febb3c9155d7833766d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-20 13:32:41.000000000","message":"Patch Set 3:\n\nStrange but dependency https://review.opendev.org/653978/ does not work. So we probably need it merged to get updated run_tests.sh","accounts_in_message":[],"_revision_number":3},{"id":"ee90bf4c832053467b4b910c20b6f892a5901712","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-20 14:05:21.000000000","message":"Patch Set 3: Code-Review-1\n\nI\u0027m working on fixing this in opendev here: https://review.opendev.org/#/c/653957/\n\nwe can back port this later, but I rather this not the default solution because it will break all of our existing releases","accounts_in_message":[],"_revision_number":3},{"id":"f94ec7db1dd782f9ae26887d76a3e96e82b76ce5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-20 14:08:01.000000000","message":"Patch Set 3: Workflow-1","accounts_in_message":[],"_revision_number":3},{"id":"60359a78ef2bc3c22a01b0c7493196ced8d38890","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 14:15:04.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/85/653985/3/check/openstack-ansible-linters/987d35f/ : FAILURE in 4m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/85/653985/3/check/openstack-ansible-deploy-aio_metal-debian-stable/d52ee05/ : FAILURE in 12m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/85/653985/3/check/openstack-ansible-deploy-aio_metal-centos-7/a0f963f/ : FAILURE in 15m 05s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/85/653985/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/1ca6e0c/ : FAILURE in 15m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/85/653985/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/cd1b6db/ : FAILURE in 12m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/85/653985/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/4676303/ : SUCCESS in 55m 04s\n- openstack-tox-docs http://logs.openstack.org/85/653985/3/check/openstack-tox-docs/d77b61c/html/ : SUCCESS in 2m 49s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/85/653985/3/check/openstack-ansible-uw_apache-centos-7/105ca43/ : FAILURE in 6m 15s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/85/653985/3/check/openstack-ansible-uw_apache-ubuntu-xenial/8826d81/ : FAILURE in 4m 55s","accounts_in_message":[],"_revision_number":3},{"id":"59eaf35b7340dd289f8fe7c62600e9ec665bb72d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-20 15:03:50.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c68f32f298afb0f0838c7e09544105e9f36114a2","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-20 15:04:06.000000000","message":"Patch Set 4: Workflow-1","accounts_in_message":[],"_revision_number":4},{"id":"5b43bceb3056733599c3edf8bd87868fcd3392aa","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-20 16:19:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/85/653985/4/check/openstack-ansible-linters/eff365f/ : FAILURE in 4m 06s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/85/653985/4/check/openstack-ansible-deploy-aio_metal-debian-stable/68a1f79/ : SUCCESS in 1h 00m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/85/653985/4/check/openstack-ansible-deploy-aio_metal-centos-7/cbe3b48/ : SUCCESS in 1h 10m 59s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/85/653985/4/check/openstack-ansible-deploy-aio_metal-opensuse-150/3b08518/ : SUCCESS in 1h 10m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/85/653985/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/70481c6/ : SUCCESS in 56m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/85/653985/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/fead17e/ : SUCCESS in 53m 39s\n- openstack-tox-docs http://logs.openstack.org/85/653985/4/check/openstack-tox-docs/e2714d4/html/ : SUCCESS in 2m 18s\n- build-openstack-releasenotes http://logs.openstack.org/85/653985/4/check/build-openstack-releasenotes/9b4a307/html/ : SUCCESS in 2m 11s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/85/653985/4/check/openstack-ansible-uw_apache-centos-7/9c17ed2/ : FAILURE in 5m 52s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/85/653985/4/check/openstack-ansible-uw_apache-ubuntu-xenial/48ac22b/ : FAILURE in 4m 04s","accounts_in_message":[],"_revision_number":4},{"id":"354a1c141ba2e99aafd06225b6107859ecb7638c","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-04-22 16:49:33.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"137cea0a0def77e72c2e82b60305aab43996a976","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-22 18:48:20.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/85/653985/4/check/openstack-ansible-linters/0a8f984/ : FAILURE in 5m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/85/653985/4/check/openstack-ansible-deploy-aio_metal-debian-stable/1a0f586/ : SUCCESS in 1h 00m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/85/653985/4/check/openstack-ansible-deploy-aio_metal-centos-7/7e70c68/ : SUCCESS in 1h 14m 15s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/85/653985/4/check/openstack-ansible-deploy-aio_metal-opensuse-150/6c0a36e/ : SUCCESS in 1h 04m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/85/653985/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/640ea6f/ : SUCCESS in 55m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/85/653985/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e7e3082/ : SUCCESS in 54m 52s\n- openstack-tox-docs http://logs.openstack.org/85/653985/4/check/openstack-tox-docs/b54bb51/html/ : SUCCESS in 3m 28s\n- build-openstack-releasenotes http://logs.openstack.org/85/653985/4/check/build-openstack-releasenotes/ee34f06/html/ : SUCCESS in 2m 32s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/85/653985/4/check/openstack-ansible-uw_apache-centos-7/8fbfb35/ : FAILURE in 8m 11s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/85/653985/4/check/openstack-ansible-uw_apache-ubuntu-xenial/858b335/ : FAILURE in 5m 43s","accounts_in_message":[],"_revision_number":4},{"id":"f97a4925ac9a9194243a742e36df1b3ad7598fe1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-12 10:15:18.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"2d9f7ce916f836da09e72248866d28fd3e1079de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-12 11:50:11.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/85/653985/5/check/openstack-ansible-linters/9b0cae8/ : SUCCESS in 5m 34s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/85/653985/5/check/openstack-ansible-deploy-aio_metal-debian-stable/ade15a8/ : SUCCESS in 55m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/85/653985/5/check/openstack-ansible-deploy-aio_metal-centos-7/179ce5e/ : SUCCESS in 1h 14m 03s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/85/653985/5/check/openstack-ansible-deploy-aio_metal-opensuse-150/5948af6/ : SUCCESS in 1h 03m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/85/653985/5/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/986cc1f/ : SUCCESS in 55m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/85/653985/5/check/openstack-ansible-deploy-aio_distro_metal-centos-7/0248943/ : FAILURE in 35m 34s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/85/653985/5/check/openstack-tox-docs/eda6cb2/html/ : SUCCESS in 2m 34s\n- build-openstack-releasenotes http://logs.openstack.org/85/653985/5/check/build-openstack-releasenotes/3ac1f48/html/ : SUCCESS in 2m 43s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/85/653985/5/check/openstack-ansible-uw_apache-centos-7/e214a49/ : SUCCESS in 37m 48s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/85/653985/5/check/openstack-ansible-uw_apache-ubuntu-xenial/82661d5/ : SUCCESS in 32m 58s","accounts_in_message":[],"_revision_number":5},{"id":"ba2590a3375a0f27ee80fd3c795e5d2ae9615d1f","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-12 16:03:17.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"9efdc452d41b26f29b4f230e90fd9c089765bb05","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-05-20 20:21:08.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"eb610df7b8a3f4019836d0749494f87427ed8422","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2019-05-20 20:37:42.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"72c110e140e7d384e628e80a4c0914922f4a6cb2","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-20 21:10:44.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5459376f56689ccee48816cd7e94b9b5e9a2555f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 21:10:56.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"221e27d1cc16101547c600e149c36e2153a7c1b5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 22:33:49.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/85/653985/5/gate/openstack-ansible-linters/ae9138d/ : SUCCESS in 4m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/85/653985/5/gate/openstack-ansible-deploy-aio_metal-debian-stable/175815e/ : SUCCESS in 1h 00m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/85/653985/5/gate/openstack-ansible-deploy-aio_metal-centos-7/fe84119/ : SUCCESS in 1h 17m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/85/653985/5/gate/openstack-ansible-deploy-aio_metal-opensuse-150/2c16e29/ : SUCCESS in 1h 06m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/85/653985/5/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e2c2d6b/ : SUCCESS in 1h 11m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/85/653985/5/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/d4dc322/ : SUCCESS in 54m 47s\n- openstack-tox-docs http://logs.openstack.org/85/653985/5/gate/openstack-tox-docs/62e2446/html/ : SUCCESS in 2m 42s\n- build-openstack-releasenotes http://logs.openstack.org/85/653985/5/gate/build-openstack-releasenotes/b59ffa6/html/ : SUCCESS in 2m 49s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/85/653985/5/gate/openstack-ansible-uw_apache-centos-7/1ac1bfa/ : SUCCESS in 40m 31s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/85/653985/5/gate/openstack-ansible-uw_apache-ubuntu-xenial/5e86399/ : FAILURE in 10m 02s","accounts_in_message":[],"_revision_number":5},{"id":"40a1ea53f2ecf0c41af7a9010a7d77c5c0721950","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-21 06:19:28.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"1a4c2e2cc29fbbd320c8871991c9243b854f43d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-21 08:08:31.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/85/653985/5/check/openstack-ansible-linters/74545f4/ : SUCCESS in 8m 24s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/85/653985/5/check/openstack-ansible-deploy-aio_metal-debian-stable/74baa1e/ : SUCCESS in 57m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/85/653985/5/check/openstack-ansible-deploy-aio_metal-centos-7/506d183/ : SUCCESS in 1h 11m 14s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/85/653985/5/check/openstack-ansible-deploy-aio_metal-opensuse-150/8d49da9/ : SUCCESS in 1h 04m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/85/653985/5/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/19c46c7/ : SUCCESS in 1h 05m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/85/653985/5/check/openstack-ansible-deploy-aio_distro_metal-centos-7/a124331/ : SUCCESS in 1h 05m 05s\n- openstack-tox-docs http://logs.openstack.org/85/653985/5/check/openstack-tox-docs/d3fcf5b/html/ : SUCCESS in 2m 53s\n- build-openstack-releasenotes http://logs.openstack.org/85/653985/5/check/build-openstack-releasenotes/7ce40fb/html/ : SUCCESS in 3m 56s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/85/653985/5/check/openstack-ansible-uw_apache-centos-7/5383489/ : SUCCESS in 36m 25s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/85/653985/5/check/openstack-ansible-uw_apache-ubuntu-xenial/debce9f/ : FAILURE in 12m 09s","accounts_in_message":[],"_revision_number":5},{"id":"3a6cd8174a86f84e353551eaa3559aa1c5596c67","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-24 17:12:37.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"0a061248265d4c961ece15eb9c9ced8ab8aac234","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-24 17:30:36.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"20145ed9859d140ded82a1f76111fcc07c4a9f59","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 18:44:08.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/85/653985/6/check/openstack-ansible-linters/e5c5085/ : SUCCESS in 5m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/85/653985/6/check/openstack-ansible-deploy-aio_metal-debian-stable/d63b867/ : SUCCESS in 58m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/85/653985/6/check/openstack-ansible-deploy-aio_metal-centos-7/cc83b7c/ : SUCCESS in 1h 16m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/85/653985/6/check/openstack-ansible-deploy-aio_metal-opensuse-150/6fd1637/ : SUCCESS in 1h 06m 31s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/85/653985/6/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f43b0f8/ : SUCCESS in 54m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/85/653985/6/check/openstack-ansible-deploy-aio_distro_metal-centos-7/1a1c437/ : SUCCESS in 56m 15s\n- openstack-tox-docs http://logs.openstack.org/85/653985/6/check/openstack-tox-docs/801682f/html/ : SUCCESS in 3m 08s\n- build-openstack-releasenotes http://logs.openstack.org/85/653985/6/check/build-openstack-releasenotes/f0637f3/html/ : SUCCESS in 2m 47s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/85/653985/6/check/openstack-ansible-uw_apache-centos-7/c987de6/ : SUCCESS in 44m 24s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/85/653985/6/check/openstack-ansible-uw_apache-ubuntu-bionic/fb15efc/ : SUCCESS in 29m 36s","accounts_in_message":[],"_revision_number":6},{"id":"06e39fd56da138c11b7a1ad34a629fa67434a8f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 18:44:21.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"ca0a9c255c46b1f2b7069ce366416af0cbb90729","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:11:03.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/85/653985/6/gate/openstack-ansible-linters/70c18e5/ : SUCCESS in 5m 26s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/85/653985/6/gate/openstack-ansible-deploy-aio_metal-debian-stable/fa84b9f/ : SUCCESS in 57m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/85/653985/6/gate/openstack-ansible-deploy-aio_metal-centos-7/fb206bf/ : SUCCESS in 1h 18m 54s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/85/653985/6/gate/openstack-ansible-deploy-aio_metal-opensuse-150/3c74658/ : SUCCESS in 1h 05m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/85/653985/6/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9de229a/ : SUCCESS in 59m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/85/653985/6/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/d977359/ : SUCCESS in 59m 31s\n- openstack-tox-docs http://logs.openstack.org/85/653985/6/gate/openstack-tox-docs/9eba571/html/ : SUCCESS in 2m 27s\n- build-openstack-releasenotes http://logs.openstack.org/85/653985/6/gate/build-openstack-releasenotes/bd1731f/html/ : SUCCESS in 2m 33s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/85/653985/6/gate/openstack-ansible-uw_apache-centos-7/3b1b558/ : SUCCESS in 49m 47s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/85/653985/6/gate/openstack-ansible-uw_apache-ubuntu-bionic/edaf0e0/ : SUCCESS in 31m 40s","accounts_in_message":[],"_revision_number":6},{"id":"9b826a3ed6d3cc2996396c36fdcf11217ebc82a4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:11:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c17062f3b4ffb649094de0a94c7bce258079baa3","revisions":{"bab731c93dad281710777429ebca7198ed7c18fd":{"kind":"REWORK","_number":1,"created":"2019-04-20 12:19:24.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/85/653985/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/85/653985/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/1"}}},"commit":{"parents":[{"commit":"ac31cf3b3defa4d9da573a539039fbd14c65afdd","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ac31cf3b3defa4d9da573a539039fbd14c65afdd"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-20 12:19:11.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-20 12:19:11.000000000","tz":180},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\n\nChange-Id: I6a6660bd3e15908adad9cd1fd76f582506562a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/bab731c93dad281710777429ebca7198ed7c18fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/bab731c93dad281710777429ebca7198ed7c18fd"}]},"branch":"refs/heads/master"},"e363cb52a2dba5270db7ef38a5583b2b03f18cd8":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-04-20 12:45:13.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/85/653985/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/85/653985/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/2"}}},"commit":{"parents":[{"commit":"ac31cf3b3defa4d9da573a539039fbd14c65afdd","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ac31cf3b3defa4d9da573a539039fbd14c65afdd"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-20 12:19:11.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-20 12:45:01.000000000","tz":180},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\n\nDepends-On: https://review.opendev.org/653970\nChange-Id: I6a6660bd3e15908adad9cd1fd76f582506562a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e363cb52a2dba5270db7ef38a5583b2b03f18cd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e363cb52a2dba5270db7ef38a5583b2b03f18cd8"}]},"branch":"refs/heads/master"},"d322fe5f3563fb7f7f040c8e1ae8f9747a69f7a8":{"kind":"REWORK","_number":3,"created":"2019-04-20 13:14:39.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/85/653985/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/85/653985/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/3"}}},"commit":{"parents":[{"commit":"ac31cf3b3defa4d9da573a539039fbd14c65afdd","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ac31cf3b3defa4d9da573a539039fbd14c65afdd"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-20 13:14:26.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-20 13:14:26.000000000","tz":180},"subject":"Fix _upper_constraints_url","message":"Fix _upper_constraints_url\n\nAfter redirect from git.openstack.org to opendev.org or method of\nretrieving upper_constraints got broken. This patch aims to fix that.\nThis is minimalistic patch that is designed to be backported.\n\nDepends-On: https://review.opendev.org/653978/\nChange-Id: I6a6660bd3e15908adad9cd1fd76f582506562a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d322fe5f3563fb7f7f040c8e1ae8f9747a69f7a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d322fe5f3563fb7f7f040c8e1ae8f9747a69f7a8"}]},"branch":"refs/heads/master"},"f51c90bcafb75be5811272d6e7b5d7b6b9bb537d":{"kind":"REWORK","_number":4,"created":"2019-04-20 15:03:50.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/85/653985/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/85/653985/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/4"}}},"commit":{"parents":[{"commit":"ac31cf3b3defa4d9da573a539039fbd14c65afdd","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ac31cf3b3defa4d9da573a539039fbd14c65afdd"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-20 15:03:39.000000000","tz":180},"committer":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-20 15:03:39.000000000","tz":180},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\n\nChange-Id: I6a6660bd3e15908adad9cd1fd76f582506562a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f51c90bcafb75be5811272d6e7b5d7b6b9bb537d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f51c90bcafb75be5811272d6e7b5d7b6b9bb537d"}]},"branch":"refs/heads/master"},"31f5bb4415cbd392f0026afd171d19f39e748631":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2019-05-12 10:15:18.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/85/653985/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/85/653985/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/5"}}},"commit":{"parents":[{"commit":"b1a0c84142fed526e20165aa0b35b16f43ac8e62","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b1a0c84142fed526e20165aa0b35b16f43ac8e62"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-20 15:03:39.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-05-12 10:15:18.000000000","tz":0},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\n\nChange-Id: I6a6660bd3e15908adad9cd1fd76f582506562a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/31f5bb4415cbd392f0026afd171d19f39e748631"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/31f5bb4415cbd392f0026afd171d19f39e748631"}]},"branch":"refs/heads/master"},"c17062f3b4ffb649094de0a94c7bce258079baa3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2019-05-24 17:12:37.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/85/653985/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/85/653985/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/85/653985/6"}}},"commit":{"parents":[{"commit":"d6d3bda11317c443a612344a9e5651e8dd26275d","subject":"Update uw_apache to run against bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d6d3bda11317c443a612344a9e5651e8dd26275d"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"noonedeadpunk@ya.ru","date":"2019-04-20 15:03:39.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2019-05-24 17:12:37.000000000","tz":0},"subject":"Replace git.openstack.org with opendev.org","message":"Replace git.openstack.org with opendev.org\n\nThis patch replaces git.openstack.org with opendev.org as redirection\nfrom old path was enabled.\n\nChange-Id: I6a6660bd3e15908adad9cd1fd76f582506562a97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c17062f3b4ffb649094de0a94c7bce258079baa3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c17062f3b4ffb649094de0a94c7bce258079baa3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
