)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~655500","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Icb8e51f2a5ec31c3ac3230733908ee8849414d94","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"bug/1826250","hashtags":[],"change_id":"Icb8e51f2a5ec31c3ac3230733908ee8849414d94","subject":"Adding keystone_service_timeout","status":"ABANDONED","created":"2019-04-24 17:10:31.000000000","updated":"2019-07-30 19:46:16.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6eb2c934e8c1c5d04174d8c2f53e210c402d0da1","_number":655500,"virtual_id_number":655500,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"date":"2019-04-26 16:37:09.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-04-25 00:42:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":-1,"date":"2019-04-26 17:34:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-04-25 00:42:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-04-26 16:37:09.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-04-26 17:34:42.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"}],"messages":[{"id":"bf4571069fe22f5e22e38319b550cf4307b20c39","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-04-24 17:10:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05e6769335dd889b00deabbd0c51d57e28cfeaba","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-24 18:35:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/00/655500/1/check/openstack-ansible-linters/68037d0/ : SUCCESS in 6m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/00/655500/1/check/openstack-ansible-deploy-aio_metal-debian-stable/82266a5/ : SUCCESS in 58m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/00/655500/1/check/openstack-ansible-deploy-aio_metal-centos-7/6ad7738/ : SUCCESS in 1h 19m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/00/655500/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/78e757d/ : SUCCESS in 1h 04m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/00/655500/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/442db23/ : SUCCESS in 57m 06s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/00/655500/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e531a23/ : SUCCESS in 56m 19s\n- openstack-tox-docs http://logs.openstack.org/00/655500/1/check/openstack-tox-docs/7ce22c8/html/ : SUCCESS in 2m 54s\n- build-openstack-releasenotes http://logs.openstack.org/00/655500/1/check/build-openstack-releasenotes/87e19e9/html/ : SUCCESS in 2m 29s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/00/655500/1/check/openstack-ansible-uw_apache-centos-7/4d6387a/ : SUCCESS in 37m 55s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/00/655500/1/check/openstack-ansible-uw_apache-ubuntu-xenial/8d3ce09/ : SUCCESS in 41m 12s","accounts_in_message":[],"_revision_number":1},{"id":"db38bb3e95f871c8b5a0c2ef581fbdf03afd7893","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-04-24 20:07:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"758c2719cd7f7fe787a762187410e46aff487eb0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-04-25 00:42:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/00/655500/2/check/openstack-ansible-linters/283c240/ : SUCCESS in 5m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/00/655500/2/check/openstack-ansible-deploy-aio_metal-debian-stable/30c72e0/ : SUCCESS in 58m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/00/655500/2/check/openstack-ansible-deploy-aio_metal-centos-7/69f8082/ : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/00/655500/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/7d420dc/ : SUCCESS in 1h 01m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/00/655500/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/806267a/ : SUCCESS in 53m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/00/655500/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/830247a/ : SUCCESS in 1h 00m 03s\n- openstack-tox-docs http://logs.openstack.org/00/655500/2/check/openstack-tox-docs/64793f2/html/ : SUCCESS in 2m 45s\n- build-openstack-releasenotes http://logs.openstack.org/00/655500/2/check/build-openstack-releasenotes/f733d55/html/ : SUCCESS in 2m 44s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/00/655500/2/check/openstack-ansible-uw_apache-centos-7/f56f765/ : SUCCESS in 37m 17s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/00/655500/2/check/openstack-ansible-uw_apache-ubuntu-xenial/f785a15/ : SUCCESS in 32m 20s","accounts_in_message":[],"_revision_number":2},{"id":"e3b10658f00a703cbce0d7cf5a5096164d12620b","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-04-26 16:37:09.000000000","message":"Patch Set 2:\n\nsee comment in https://review.opendev.org/#/c/655526/2","accounts_in_message":[],"_revision_number":2},{"id":"4278a18f37d89741ad8a3c70e28e86754fe5740a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-04-26 17:34:42.000000000","message":"Patch Set 2: Code-Review-1\n\nplease review feedback on https://review.opendev.org/#/c/655559/1","accounts_in_message":[],"_revision_number":2},{"id":"ebb0b507eec1431cbde1a620eeed58fe10c91a0b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-30 19:46:16.000000000","message":"Abandoned\n\nno follow up patch and adds an unnecessary variable which can just be used by an override","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fbeb131b2188b5d0c67917fd54bf5c7cb06c9610","revisions":{"7ad981c14fdfbf091611ff317d5823fa0872f345":{"kind":"REWORK","_number":1,"created":"2019-04-24 17:10:31.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/00/655500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/00/655500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/655500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/655500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/655500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/655500/1"}}},"commit":{"parents":[{"commit":"c3f1866e41808ece417c7a3dbb84d74ea5c9511c","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c3f1866e41808ece417c7a3dbb84d74ea5c9511c"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 17:02:02.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 17:07:16.000000000","tz":-300},"subject":"Adding keystone_service_timeout","message":"Adding keystone_service_timeout\n\nThe override `keystone_service_timeout` is introduced to allow\nmultiple service components to use the same frontend/backend\ntimeouts. The default value is 60 (seconds).\n\nCloses-Bug: 1826250\n\nChange-Id: Icb8e51f2a5ec31c3ac3230733908ee8849414d94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7ad981c14fdfbf091611ff317d5823fa0872f345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7ad981c14fdfbf091611ff317d5823fa0872f345"}]},"branch":"refs/heads/master"},"fbeb131b2188b5d0c67917fd54bf5c7cb06c9610":{"kind":"REWORK","_number":2,"created":"2019-04-24 20:07:18.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/00/655500/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/00/655500/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/655500/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/655500/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/655500/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/655500/2"}}},"commit":{"parents":[{"commit":"c3f1866e41808ece417c7a3dbb84d74ea5c9511c","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c3f1866e41808ece417c7a3dbb84d74ea5c9511c"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 17:02:02.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-04-24 20:07:09.000000000","tz":-300},"subject":"Adding keystone_service_timeout","message":"Adding keystone_service_timeout\n\nThe override `keystone_service_timeout` is introduced to allow\nmultiple service components to use the same frontend/backend\ntimeouts. The default value is 60 (seconds).\n\nCloses-Bug: 1826250\n\nChange-Id: Icb8e51f2a5ec31c3ac3230733908ee8849414d94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/fbeb131b2188b5d0c67917fd54bf5c7cb06c9610"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/fbeb131b2188b5d0c67917fd54bf5c7cb06c9610"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
