)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~659450","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Iae8cc8df4bf91b13e3746d82d0cb751d18e35945","project":"openstack/openstack-ansible-os_keystone","branch":"master","hashtags":[],"change_id":"Iae8cc8df4bf91b13e3746d82d0cb751d18e35945","subject":"Cap sphinx for py2 to match global requirements","status":"MERGED","created":"2019-05-16 06:42:03.000000000","updated":"2019-05-19 07:31:05.000000000","submitted":"2019-05-19 07:31:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"659450-1558251065268-334d6ac5","meta_rev_id":"8248a6db0926067f0cab791a356d61b5e3103573","_number":659450,"virtual_id_number":659450,"owner":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-19 07:31:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-05-18 11:21:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-19 07:23:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-05-19 07:23:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-18 11:21:04.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-19 07:23:30.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-19 07:31:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"909505b93e5795e66e14ebc2b776c0b6b2f53f4e","author":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"date":"2019-05-16 06:42:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d1907a07983209ead87f810763770b89a6b5dc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-16 06:54:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/50/659450/1/check/requirements-check/3551ed7/ : SUCCESS in 2m 04s\n- openstack-ansible-linters http://logs.openstack.org/50/659450/1/check/openstack-ansible-linters/b2b9f4f/ : SUCCESS in 5m 43s\n- openstack-tox-docs http://logs.openstack.org/50/659450/1/check/openstack-tox-docs/1bcd619/html/ : SUCCESS in 2m 36s\n- build-openstack-releasenotes http://logs.openstack.org/50/659450/1/check/build-openstack-releasenotes/ffd765f/html/ : SUCCESS in 2m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a9811e0456825cb1ab14324a29a6783fdd40bd75","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-18 11:21:04.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"587d6d1d9db0260432512055db621d7dc20b578a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-19 07:23:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a4e4cad9b161399e17bc11487134ee8520a87b31","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 07:23:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e6eaafebd0c4fa4fd78e66d4a909525eab0e64a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 07:31:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8248a6db0926067f0cab791a356d61b5e3103573","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-19 07:31:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/50/659450/1/gate/requirements-check/d612c71/ : SUCCESS in 2m 11s\n- openstack-ansible-linters http://logs.openstack.org/50/659450/1/gate/openstack-ansible-linters/b755cdb/ : SUCCESS in 5m 42s\n- openstack-tox-docs http://logs.openstack.org/50/659450/1/gate/openstack-tox-docs/30bbc2b/html/ : SUCCESS in 2m 38s\n- build-openstack-releasenotes http://logs.openstack.org/50/659450/1/gate/build-openstack-releasenotes/ac1bf1e/html/ : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3c0ee6269604401cb5e57d4ff542f6fe8bed99e3","revisions":{"3c0ee6269604401cb5e57d4ff542f6fe8bed99e3":{"kind":"REWORK","_number":1,"created":"2019-05-16 06:42:03.000000000","uploader":{"_account_id":30384,"name":"zhangboye","email":"zhangby09@chinatelecom.cn","username":"zhangboye"},"ref":"refs/changes/50/659450/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/50/659450/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/50/659450/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/50/659450/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/50/659450/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/50/659450/1"}}},"commit":{"parents":[{"commit":"6057dcbbc367918b6f86a03bb862984c63c58552","subject":"[redhat] Update variables to be version agnostic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6057dcbbc367918b6f86a03bb862984c63c58552"}]}],"author":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-05-16 06:41:35.000000000","tz":480},"committer":{"name":"zhangboye","email":"zhangboye@inspur.com","date":"2019-05-16 06:41:58.000000000","tz":480},"subject":"Cap sphinx for py2 to match global requirements","message":"Cap sphinx for py2 to match global requirements\n\nChange-Id: Iae8cc8df4bf91b13e3746d82d0cb751d18e35945\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3c0ee6269604401cb5e57d4ff542f6fe8bed99e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3c0ee6269604401cb5e57d4ff542f6fe8bed99e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
