)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~660084","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I8d4bccbe283a0b09f42e2d2419612b3a4a1c6a0d","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"clarify-suse-support","hashtags":[],"change_id":"I8d4bccbe283a0b09f42e2d2419612b3a4a1c6a0d","subject":"Clarify supported SUSE versions","status":"MERGED","created":"2019-05-20 10:38:04.000000000","updated":"2019-05-27 21:20:38.000000000","submitted":"2019-05-27 21:20:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"660084-1558992038820-dc81486c","meta_rev_id":"693feb9918f35ae7594cfb2e9c5bc02555907a91","_number":660084,"virtual_id_number":660084,"owner":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-05-27 21:20:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-05-27 07:15:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-05-27 19:46:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":1,"date":"2019-05-27 19:46:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-27 07:15:28.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-05-27 19:46:22.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-05-27 21:20:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"10d7878e0854ae7a297fff39dfa2079b35a8b287","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-05-20 10:38:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5570547e1adde388f6a8d634c3673b23a857acb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-20 12:09:38.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/84/660084/1/check/openstack-ansible-linters/605dbf6/ : SUCCESS in 4m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/84/660084/1/check/openstack-ansible-deploy-aio_metal-debian-stable/6233d64/ : SUCCESS in 58m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/84/660084/1/check/openstack-ansible-deploy-aio_metal-centos-7/79478a2/ : SUCCESS in 1h 23m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/84/660084/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/34a9ea5/ : SUCCESS in 1h 08m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/84/660084/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/56cf4f7/ : SUCCESS in 56m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/84/660084/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/323d483/ : SUCCESS in 51m 48s\n- openstack-tox-docs http://logs.openstack.org/84/660084/1/check/openstack-tox-docs/7f7cddf/html/ : SUCCESS in 3m 14s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/84/660084/1/check/openstack-ansible-uw_apache-centos-7/82fcec9/ : SUCCESS in 35m 29s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/84/660084/1/check/openstack-ansible-uw_apache-ubuntu-xenial/02e67fb/ : FAILURE in 9m 05s","accounts_in_message":[],"_revision_number":1},{"id":"adf019ce57b4c22ec4b072e529d9149374665671","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-27 07:15:23.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"cf4665569a68354df894e53ba7b33535d8dc31f1","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-05-27 07:15:28.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"25bbe1b88c4796d376101ad928d8501a4909a889","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 08:44:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/84/660084/2/check/openstack-ansible-linters/d45567e/ : SUCCESS in 5m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/84/660084/2/check/openstack-ansible-deploy-aio_metal-debian-stable/2557933/ : SUCCESS in 56m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/84/660084/2/check/openstack-ansible-deploy-aio_metal-centos-7/d7d5e00/ : SUCCESS in 1h 20m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/84/660084/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/e58c458/ : SUCCESS in 1h 06m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/84/660084/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/0bef9bb/ : SUCCESS in 54m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/84/660084/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/a889ba7/ : SUCCESS in 49m 55s\n- openstack-tox-docs http://logs.openstack.org/84/660084/2/check/openstack-tox-docs/47c4bc7/html/ : SUCCESS in 2m 33s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/84/660084/2/check/openstack-ansible-uw_apache-centos-7/1e4e333/ : SUCCESS in 38m 07s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/84/660084/2/check/openstack-ansible-uw_apache-ubuntu-bionic/4dd98b4/ : SUCCESS in 30m 47s","accounts_in_message":[],"_revision_number":2},{"id":"4897333522fbfe8166ae4a374e4efff88212e494","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-05-27 19:46:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"81e417e29c5962be5bfb5e4754c47e5f53ae2b87","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 19:46:33.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"0605d6ab057f8c98622284287c7b3ba5e5511f03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 21:20:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"2a587c5c85e6b6be840254ea7f300f9d88c534c9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 21:20:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/84/660084/2/gate/openstack-ansible-linters/e268d12/ : SUCCESS in 8m 43s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/84/660084/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/6071c1e/ : SUCCESS in 1h 01m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/84/660084/2/gate/openstack-ansible-deploy-aio_metal-centos-7/043c474/ : SUCCESS in 1h 20m 22s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/84/660084/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/0c8a833/ : SUCCESS in 1h 03m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/84/660084/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/ffde93d/ : SUCCESS in 1h 13m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/84/660084/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/3edc091/ : SUCCESS in 1h 08m 20s\n- openstack-tox-docs http://logs.openstack.org/84/660084/2/gate/openstack-tox-docs/eef8d89/html/ : SUCCESS in 3m 46s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/84/660084/2/gate/openstack-ansible-uw_apache-centos-7/d9a48d0/ : SUCCESS in 50m 33s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/84/660084/2/gate/openstack-ansible-uw_apache-ubuntu-bionic/de20417/ : SUCCESS in 41m 26s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"dfdb5408046fd1c94f43b7748a8679771c2bfd33","revisions":{"4ce7ad3446d3134d469de282df4ae2c2f72dde1b":{"kind":"REWORK","_number":1,"created":"2019-05-20 10:38:04.000000000","uploader":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"ref":"refs/changes/84/660084/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/84/660084/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/84/660084/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/84/660084/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/84/660084/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/84/660084/1"}}},"commit":{"parents":[{"commit":"3c0ee6269604401cb5e57d4ff542f6fe8bed99e3","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3c0ee6269604401cb5e57d4ff542f6fe8bed99e3"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-20 10:37:05.000000000","tz":120},"committer":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-20 10:37:46.000000000","tz":120},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: I8d4bccbe283a0b09f42e2d2419612b3a4a1c6a0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/4ce7ad3446d3134d469de282df4ae2c2f72dde1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/4ce7ad3446d3134d469de282df4ae2c2f72dde1b"}]},"branch":"refs/heads/master"},"dfdb5408046fd1c94f43b7748a8679771c2bfd33":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-05-27 07:15:23.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/84/660084/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/84/660084/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/84/660084/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/84/660084/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/84/660084/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/84/660084/2"}}},"commit":{"parents":[{"commit":"d6d3bda11317c443a612344a9e5651e8dd26275d","subject":"Update uw_apache to run against bionic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d6d3bda11317c443a612344a9e5651e8dd26275d"}]}],"author":{"name":"Jean-Philippe Evrard","email":"jean-philippe@evrard.me","date":"2019-05-20 10:37:05.000000000","tz":120},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2019-05-27 07:15:23.000000000","tz":0},"subject":"Clarify supported SUSE versions","message":"Clarify supported SUSE versions\n\nThis role only support openSUSE Leap 15, not 42.3, not tumbleweed.\nSaying all is confusing and a bad practice. This fixes it, while\nensuring the job is properly defined to test that assertion.\n\nChange-Id: I8d4bccbe283a0b09f42e2d2419612b3a4a1c6a0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/dfdb5408046fd1c94f43b7748a8679771c2bfd33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/dfdb5408046fd1c94f43b7748a8679771c2bfd33"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
