)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~661080","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I5117ccb6395c820a19d9070187b0a1e5c9fba448","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"cleanup-wsgi-apache","hashtags":[],"change_id":"I5117ccb6395c820a19d9070187b0a1e5c9fba448","subject":"remove old wsgi script in apache","status":"MERGED","created":"2019-05-23 18:31:05.000000000","updated":"2019-06-04 15:56:00.000000000","submitted":"2019-05-27 17:35:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661080-1558978557299-72d99f86","meta_rev_id":"6c1833fc0a7b5ad2c1c7fbd301577302edda74d6","_number":661080,"virtual_id_number":661080,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-05-27 17:35:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2019-05-27 15:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-05-27 16:18:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-05-27 16:18:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-27 15:55:40.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-05-27 16:18:40.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-05-27 17:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5257c5a1d56b8e728cc855cc71fb326f1e17f876","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-05-23 18:31:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"891f7086af27ba99b6fb8f47aead3b15d94b801b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-23 19:55:32.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/80/661080/1/check/openstack-ansible-linters/1e253fe/ : SUCCESS in 5m 14s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/80/661080/1/check/openstack-ansible-deploy-aio_metal-debian-stable/f2693bd/ : SUCCESS in 59m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/80/661080/1/check/openstack-ansible-deploy-aio_metal-centos-7/54f0544/ : SUCCESS in 1h 17m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/80/661080/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/1c56111/ : FAILURE in 15m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/80/661080/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/adcadce/ : SUCCESS in 54m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/80/661080/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/babea30/ : SUCCESS in 58m 58s\n- openstack-tox-docs http://logs.openstack.org/80/661080/1/check/openstack-tox-docs/4dabcce/html/ : SUCCESS in 2m 47s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/80/661080/1/check/openstack-ansible-uw_apache-centos-7/fc4eb84/ : SUCCESS in 41m 24s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/80/661080/1/check/openstack-ansible-uw_apache-ubuntu-xenial/d93675b/ : FAILURE in 12m 13s","accounts_in_message":[],"_revision_number":1},{"id":"eef893b3cd0a45c74a2ff11c2822477572920cdc","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-05-24 15:17:24.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"88021225b83f758b61d4d7abaa50a956699bacf2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 16:56:48.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/80/661080/2/check/openstack-ansible-linters/fe65aee/ : SUCCESS in 5m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/80/661080/2/check/openstack-ansible-deploy-aio_metal-debian-stable/133d81a/ : SUCCESS in 56m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/80/661080/2/check/openstack-ansible-deploy-aio_metal-centos-7/8d11f94/ : SUCCESS in 1h 22m 52s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/80/661080/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/88caa12/ : SUCCESS in 1h 12m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/80/661080/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9ffdad5/ : FAILURE in 26m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/80/661080/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/35c428c/ : SUCCESS in 52m 54s\n- openstack-tox-docs http://logs.openstack.org/80/661080/2/check/openstack-tox-docs/5c7bca5/html/ : SUCCESS in 2m 33s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/80/661080/2/check/openstack-ansible-uw_apache-centos-7/29e9949/ : SUCCESS in 39m 26s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/80/661080/2/check/openstack-ansible-uw_apache-ubuntu-bionic/8e721fa/ : SUCCESS in 30m 00s","accounts_in_message":[],"_revision_number":2},{"id":"623314d17a617848a0b177167713b00b7d6ce5fc","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-05-24 17:23:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6706e026bca0da38ba79840b53e7cbca39a132e0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 19:00:50.000000000","message":"Patch Set 2:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":2},{"id":"6b34db9834e8df1beb5c9423d272dca55f8466c8","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-05-27 12:06:36.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"8280e31355e796e18de40b5bab1071e3a094ac29","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-05-27 12:06:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fcdf3c04ac51e990322e6e2ccd1cc0cf53d130bd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 13:32:06.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/80/661080/3/check/openstack-ansible-linters/264b561/ : SUCCESS in 5m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/80/661080/3/check/openstack-ansible-deploy-aio_metal-debian-stable/8f4d042/ : SUCCESS in 1h 03m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/80/661080/3/check/openstack-ansible-deploy-aio_metal-centos-7/e6529c4/ : SUCCESS in 1h 19m 41s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/80/661080/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/5cebac3/ : SUCCESS in 1h 07m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/80/661080/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/e20d0a9/ : SUCCESS in 53m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/80/661080/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/42d06f3/ : SUCCESS in 52m 11s\n- openstack-tox-docs http://logs.openstack.org/80/661080/3/check/openstack-tox-docs/dfd6c88/html/ : SUCCESS in 2m 35s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/80/661080/3/check/openstack-ansible-uw_apache-centos-7/cc9bc10/ : SUCCESS in 35m 32s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/80/661080/3/check/openstack-ansible-uw_apache-ubuntu-bionic/f563c43/ : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":3},{"id":"15dbd94506a228ec00196aeebfe8e8e93fc8f91f","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-05-27 15:55:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"4cdbcbecca20d8b396238a564453114dcf3e1942","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-05-27 16:18:40.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"165e90b568ebf668ff44a98af6affbf5370f7294","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 16:18:52.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"2e42dfacb736924502bf1aa0abbf80ab18e2688f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 17:35:57.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/80/661080/3/gate/openstack-ansible-linters/c4accdc/ : SUCCESS in 4m 16s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/80/661080/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/3bc741a/ : SUCCESS in 53m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/80/661080/3/gate/openstack-ansible-deploy-aio_metal-centos-7/ddb77bd/ : SUCCESS in 1h 11m 49s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/80/661080/3/gate/openstack-ansible-deploy-aio_metal-opensuse-150/492781b/ : SUCCESS in 1h 07m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/80/661080/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/6a386c9/ : SUCCESS in 56m 34s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/80/661080/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/bf31a47/ : SUCCESS in 52m 42s\n- openstack-tox-docs http://logs.openstack.org/80/661080/3/gate/openstack-tox-docs/8f73189/html/ : SUCCESS in 2m 29s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/80/661080/3/gate/openstack-ansible-uw_apache-centos-7/57eafe7/ : SUCCESS in 36m 31s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/80/661080/3/gate/openstack-ansible-uw_apache-ubuntu-bionic/45bc404/ : SUCCESS in 30m 08s","accounts_in_message":[],"_revision_number":3},{"id":"918ca41f58a067f72b3e6c38989bd6362fe5f139","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 17:35:57.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3b096d18bd8dc177f43b7f09a2901313faf7b3d1","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-04 15:55:51.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 0e18ca5d1062c2e3bf37b5f0a0d1f81e7636ff9a","accounts_in_message":[],"_revision_number":3},{"id":"023d820a793c3d0272bc612be1a669e72ea08b64","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-06-04 15:56:00.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 5ba009047905a10af4dd1c0208bb101ec4792e2f","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"fbed9e974c4198c18216ca4ad6297d438ba3256e","revisions":{"3e8e367d8b99ca7d538014b0f76bfe90a968cc46":{"kind":"REWORK","_number":1,"created":"2019-05-23 18:31:05.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/80/661080/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/80/661080/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/80/661080/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/80/661080/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/80/661080/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/80/661080/1"}}},"commit":{"parents":[{"commit":"3c0ee6269604401cb5e57d4ff542f6fe8bed99e3","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3c0ee6269604401cb5e57d4ff542f6fe8bed99e3"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2019-05-23 18:25:28.000000000","tz":-240},"committer":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2019-05-23 18:25:28.000000000","tz":-240},"subject":"remove old wsgi script in apache","message":"remove old wsgi script in apache\n\nWSGIScriptAliasMatch is a configuration for mod_wsgi that is not used\nanymore.\nPrevent ProxyPass to redirect all traffic to keystone uwsgi, the\n/Shibboleth.sso needs to be handled by mod_shib\n\nChange-Id: I5117ccb6395c820a19d9070187b0a1e5c9fba448\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3e8e367d8b99ca7d538014b0f76bfe90a968cc46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3e8e367d8b99ca7d538014b0f76bfe90a968cc46"}]},"branch":"refs/heads/master"},"b864337fb782786dd10c3d5ad4103878396bb361":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-05-24 15:17:24.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/80/661080/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/80/661080/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/80/661080/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/80/661080/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/80/661080/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/80/661080/2"}}},"commit":{"parents":[{"commit":"3c0ee6269604401cb5e57d4ff542f6fe8bed99e3","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3c0ee6269604401cb5e57d4ff542f6fe8bed99e3"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2019-05-23 18:25:28.000000000","tz":-240},"committer":{"name":"Marc Gariépy (mgariepy)","email":"gariepy.marc@gmail.com","date":"2019-05-24 15:17:20.000000000","tz":0},"subject":"remove old wsgi script in apache","message":"remove old wsgi script in apache\n\nWSGIScriptAliasMatch is a configuration for mod_wsgi that is not used\nanymore.\nPrevent ProxyPass to redirect all traffic to keystone uwsgi, the\n/Shibboleth.sso needs to be handled by mod_shib\n\nChange-Id: I5117ccb6395c820a19d9070187b0a1e5c9fba448\nDepends-On: Ibff3aa2a1ac1bbf2493aaf2419ee1e4dd763934c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b864337fb782786dd10c3d5ad4103878396bb361"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b864337fb782786dd10c3d5ad4103878396bb361"}]},"branch":"refs/heads/master"},"fbed9e974c4198c18216ca4ad6297d438ba3256e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2019-05-27 12:06:53.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/80/661080/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/80/661080/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/80/661080/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/80/661080/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/80/661080/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/80/661080/3"}}},"commit":{"parents":[{"commit":"c17062f3b4ffb649094de0a94c7bce258079baa3","subject":"Replace git.openstack.org with opendev.org","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c17062f3b4ffb649094de0a94c7bce258079baa3"}]}],"author":{"name":"Marc Gariepy","email":"gariepy.marc@gmail.com","date":"2019-05-23 18:25:28.000000000","tz":-240},"committer":{"name":"Marc Gariépy (mgariepy)","email":"gariepy.marc@gmail.com","date":"2019-05-27 12:06:53.000000000","tz":0},"subject":"remove old wsgi script in apache","message":"remove old wsgi script in apache\n\nWSGIScriptAliasMatch is a configuration for mod_wsgi that is not used\nanymore.\nPrevent ProxyPass to redirect all traffic to keystone uwsgi, the\n/Shibboleth.sso needs to be handled by mod_shib\n\nChange-Id: I5117ccb6395c820a19d9070187b0a1e5c9fba448\nDepends-On: Ibff3aa2a1ac1bbf2493aaf2419ee1e4dd763934c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/fbed9e974c4198c18216ca4ad6297d438ba3256e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/fbed9e974c4198c18216ca4ad6297d438ba3256e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
