)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~661333","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ifec88251712a99ad7afbb51e386adf15bc4841eb","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"bug/1830397","hashtags":[],"change_id":"Ifec88251712a99ad7afbb51e386adf15bc4841eb","subject":"Prevent autoremoval of packages on bare metal","status":"MERGED","created":"2019-05-24 15:26:14.000000000","updated":"2019-06-11 22:45:09.000000000","submitted":"2019-06-11 22:45:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661333-1560293109107-8840775f","meta_rev_id":"96f572b89638e66f7cdb18aedeff62f8e4fd3b0d","_number":661333,"virtual_id_number":661333,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-11 22:45:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":2,"date":"2019-06-11 18:17:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":2,"date":"2019-06-11 21:13:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":1,"date":"2019-06-11 21:13:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-24 16:42:17.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-05 15:36:13.000000000","updated_by":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"reviewer":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"state":"REVIEWER"},{"updated":"2019-06-05 20:26:18.000000000","updated_by":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"reviewer":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"state":"REVIEWER"},{"updated":"2019-06-11 10:19:30.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-11 18:17:14.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-11 21:13:27.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-11 22:45:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c29651e9d1869da8e3b1c17217ebb83b9e17d92","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-05-24 15:26:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70fe4e2f380dfcf6ce60019e6bc7a2bb450819b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 17:19:07.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/33/661333/1/check/openstack-ansible-linters/d838e12/ : SUCCESS in 5m 59s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/33/661333/1/check/openstack-ansible-deploy-aio_metal-debian-stable/409e235/ : SUCCESS in 58m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/33/661333/1/check/openstack-ansible-deploy-aio_metal-centos-7/ba16bb6/ : SUCCESS in 1h 33m 31s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/33/661333/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/7d90f88/ : SUCCESS in 59m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/33/661333/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/22857d2/ : SUCCESS in 59m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/33/661333/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/9f14f92/ : SUCCESS in 51m 19s\n- openstack-tox-docs http://logs.openstack.org/33/661333/1/check/openstack-tox-docs/c6bd537/html/ : SUCCESS in 2m 54s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/33/661333/1/check/openstack-ansible-uw_apache-centos-7/f6ad2dd/ : FAILURE in 19m 51s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/33/661333/1/check/openstack-ansible-uw_apache-ubuntu-xenial/727bbd3/ : FAILURE in 21m 43s","accounts_in_message":[],"_revision_number":1},{"id":"a32274fb57196e0e9507740d900bc1b60f202491","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-05-29 18:59:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a30061b4b3cdbdced11c1bd01f7c06e468220788","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-29 20:17:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/33/661333/2/check/openstack-ansible-linters/ec45376/ : SUCCESS in 5m 21s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/33/661333/2/check/openstack-ansible-deploy-aio_metal-debian-stable/9b15335/ : SUCCESS in 56m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/33/661333/2/check/openstack-ansible-deploy-aio_metal-centos-7/490d2f8/ : SUCCESS in 1h 12m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/33/661333/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/a7bb415/ : SUCCESS in 1h 09m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/33/661333/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b0e5601/ : SUCCESS in 53m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/33/661333/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/c3a840f/ : SUCCESS in 57m 24s\n- openstack-tox-docs http://logs.openstack.org/33/661333/2/check/openstack-tox-docs/8bdc791/html/ : SUCCESS in 2m 36s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/33/661333/2/check/openstack-ansible-uw_apache-centos-7/03da1e6/ : SUCCESS in 42m 41s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/33/661333/2/check/openstack-ansible-uw_apache-ubuntu-bionic/a28793e/ : SUCCESS in 31m 22s","accounts_in_message":[],"_revision_number":2},{"id":"06e16b566ff9a43b2277befa71d8805758575b9d","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-06-04 03:00:09.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"70967a780632f6c4ac20057b573ec8949605a281","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-05 13:41:34.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"43573c329eb276a4ed8d6e2080b469b22e4961ff","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-06-05 15:35:48.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0633e087d3fe98e68693c16b0a20718f7845fb4a","author":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"date":"2019-06-05 15:36:13.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"bf4adf4f29bafda9e7de8ae9079969d66aef29ae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-05 17:13:25.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/33/661333/3/check/openstack-ansible-linters/59e9832/ : SUCCESS in 9m 46s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/33/661333/3/check/openstack-ansible-deploy-aio_metal-debian-stable/e86cc49/ : SUCCESS in 1h 04m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/33/661333/3/check/openstack-ansible-deploy-aio_metal-centos-7/f5493f7/ : SUCCESS in 1h 24m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/33/661333/3/check/openstack-ansible-deploy-aio_metal-opensuse-150/9fad708/ : SUCCESS in 1h 09m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/33/661333/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/bc04894/ : SUCCESS in 57m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/33/661333/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/62a9679/ : SUCCESS in 1h 08m 09s\n- openstack-tox-docs http://logs.openstack.org/33/661333/3/check/openstack-tox-docs/291e15c/html/ : SUCCESS in 2m 24s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/33/661333/3/check/openstack-ansible-uw_apache-centos-7/769f47e/ : SUCCESS in 36m 52s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/33/661333/3/check/openstack-ansible-uw_apache-ubuntu-bionic/0b4f89f/ : SUCCESS in 35m 10s","accounts_in_message":[],"_revision_number":3},{"id":"b51391b2a734271c09aec6fba25aa55a410babfa","author":{"_account_id":290,"name":"Antony Messerli","email":"antony@mes.ser.li","username":"antonym"},"date":"2019-06-05 20:26:18.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7c5d21ab3c49d33855b9fc50b365a302c776d409","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-06 01:51:51.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"31ed349381b42ad701fd4bf80656ff1004f4aad1","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-11 10:19:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f8536bd4e58f0846af80d444dda262e3b348a171","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-06-11 13:56:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c9ab58b1a3a267e721e4b647c80661992c7b7af3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 15:18:22.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/33/661333/4/check/openstack-ansible-linters/66527c7/ : SUCCESS in 6m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/33/661333/4/check/openstack-ansible-deploy-aio_metal-debian-stable/9cf1afa/ : SUCCESS in 59m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/33/661333/4/check/openstack-ansible-deploy-aio_metal-centos-7/831898b/ : SUCCESS in 1h 18m 58s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/33/661333/4/check/openstack-ansible-deploy-aio_metal-opensuse-150/8df72b3/ : SUCCESS in 1h 12m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/33/661333/4/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/1eda35f/ : SUCCESS in 56m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/33/661333/4/check/openstack-ansible-deploy-aio_distro_metal-centos-7/2e472c1/ : SUCCESS in 56m 34s\n- openstack-tox-docs http://logs.openstack.org/33/661333/4/check/openstack-tox-docs/ed12cf2/html/ : SUCCESS in 3m 16s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/33/661333/4/check/openstack-ansible-uw_apache-centos-7/b9c99ad/ : SUCCESS in 39m 18s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/33/661333/4/check/openstack-ansible-uw_apache-ubuntu-bionic/767bb6c/ : SUCCESS in 31m 21s","accounts_in_message":[],"_revision_number":4},{"id":"a33581c9df3f382d935d0b20eee29b67a6788d70","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-11 18:17:14.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9933ce3c29dbe56b38ac38217cb6b90c529fef0f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 21:13:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"8e88e51bcd8b80e5987eb37fa189be2f72dc601b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:13:41.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"662a6e20e408a1ce1cf522e16f5dea8dfb3b2c5d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:45:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/33/661333/4/gate/openstack-ansible-linters/57e3001/ : SUCCESS in 6m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/33/661333/4/gate/openstack-ansible-deploy-aio_metal-debian-stable/d8f7bc3/ : SUCCESS in 1h 07m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/33/661333/4/gate/openstack-ansible-deploy-aio_metal-centos-7/1731186/ : SUCCESS in 1h 24m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/33/661333/4/gate/openstack-ansible-deploy-aio_metal-opensuse-150/acbc489/ : SUCCESS in 1h 09m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/33/661333/4/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/0bce22d/ : SUCCESS in 58m 38s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/33/661333/4/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/5ec1438/ : SUCCESS in 59m 47s\n- openstack-tox-docs http://logs.openstack.org/33/661333/4/gate/openstack-tox-docs/f50f5be/html/ : SUCCESS in 2m 52s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/33/661333/4/gate/openstack-ansible-uw_apache-centos-7/e4c603b/ : SUCCESS in 39m 24s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/33/661333/4/gate/openstack-ansible-uw_apache-ubuntu-bionic/2c3196e/ : SUCCESS in 41m 11s","accounts_in_message":[],"_revision_number":4},{"id":"cb12274b3517488cb036c34b7fdb1acacb8e652f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 22:45:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ae6c1982ec2101547b6744cf157aae4ce5bbb49e","revisions":{"9407c8b5d53fde28e785c9e8b4925acf8fae41bd":{"kind":"REWORK","_number":1,"created":"2019-05-24 15:26:14.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/33/661333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/33/661333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/1"}}},"commit":{"parents":[{"commit":"3c0ee6269604401cb5e57d4ff542f6fe8bed99e3","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3c0ee6269604401cb5e57d4ff542f6fe8bed99e3"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-05-24 15:24:36.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-05-24 15:24:36.000000000","tz":-300},"subject":"Prevent autoremoval of packages on bare metal","message":"Prevent autoremoval of packages on bare metal\n\nWhen deploying OSA on bare metal (no container tech),\nthe os keystone role interferes with os horizon as the\npackage apache2 is automatically uninstalled.\n\nChange-Id: Ifec88251712a99ad7afbb51e386adf15bc4841eb\nCloses-Bug: #1830397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9407c8b5d53fde28e785c9e8b4925acf8fae41bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9407c8b5d53fde28e785c9e8b4925acf8fae41bd"}]},"branch":"refs/heads/master"},"258447e54fa1e9a679fcb17e33bc2fb3b7c80aa1":{"kind":"REWORK","_number":2,"created":"2019-05-29 18:59:17.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/33/661333/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/33/661333/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/2"}}},"commit":{"parents":[{"commit":"3c0ee6269604401cb5e57d4ff542f6fe8bed99e3","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3c0ee6269604401cb5e57d4ff542f6fe8bed99e3"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-05-24 15:24:36.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-05-29 18:59:09.000000000","tz":-300},"subject":"Prevent autoremoval of packages on bare metal","message":"Prevent autoremoval of packages on bare metal\n\nWhen deploying OSA on bare metal (no container tech),\nthe os keystone role interferes with os horizon as the\npackage apache2 is automatically uninstalled.\n\nChange-Id: Ifec88251712a99ad7afbb51e386adf15bc4841eb\nCloses-Bug: #1830397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/258447e54fa1e9a679fcb17e33bc2fb3b7c80aa1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/258447e54fa1e9a679fcb17e33bc2fb3b7c80aa1"}]},"branch":"refs/heads/master"},"dee552dce03a0bed1abd7b9c89f30d05cc540eb1":{"kind":"REWORK","_number":3,"created":"2019-06-05 15:35:48.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/33/661333/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/33/661333/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/3"}}},"commit":{"parents":[{"commit":"3c0ee6269604401cb5e57d4ff542f6fe8bed99e3","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3c0ee6269604401cb5e57d4ff542f6fe8bed99e3"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-05-24 15:24:36.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-06-05 15:35:42.000000000","tz":-300},"subject":"Prevent autoremoval of packages on bare metal","message":"Prevent autoremoval of packages on bare metal\n\nWhen deploying OSA on bare metal (no container tech),\nthe os keystone role interferes with os horizon as the\npackage apache2 is automatically uninstalled.\n\nChange-Id: Ifec88251712a99ad7afbb51e386adf15bc4841eb\nCloses-Bug: #1830397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/dee552dce03a0bed1abd7b9c89f30d05cc540eb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/dee552dce03a0bed1abd7b9c89f30d05cc540eb1"}]},"branch":"refs/heads/master"},"ae6c1982ec2101547b6744cf157aae4ce5bbb49e":{"kind":"REWORK","_number":4,"created":"2019-06-11 13:56:12.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/33/661333/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/33/661333/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/33/661333/4"}}},"commit":{"parents":[{"commit":"3c0ee6269604401cb5e57d4ff542f6fe8bed99e3","subject":"Cap sphinx for py2 to match global requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3c0ee6269604401cb5e57d4ff542f6fe8bed99e3"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-05-24 15:24:36.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-06-11 13:47:50.000000000","tz":-300},"subject":"Prevent autoremoval of packages on bare metal","message":"Prevent autoremoval of packages on bare metal\n\nWhen deploying OSA on bare metal (no container tech),\nthe os keystone role interferes with os horizon as the\npackage apache2 is automatically uninstalled.\n\nChange-Id: Ifec88251712a99ad7afbb51e386adf15bc4841eb\nCloses-Bug: #1830397\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ae6c1982ec2101547b6744cf157aae4ce5bbb49e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ae6c1982ec2101547b6744cf157aae4ce5bbb49e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
