)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~661378","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fstein~Ibff3aa2a1ac1bbf2493aaf2419ee1e4dd763934c","project":"openstack/openstack-ansible-os_keystone","branch":"stable/stein","topic":"uw_apache_bionic-stable/stein","hashtags":[],"change_id":"Ibff3aa2a1ac1bbf2493aaf2419ee1e4dd763934c","subject":"Update uw_apache to run against bionic","status":"MERGED","created":"2019-05-24 19:00:38.000000000","updated":"2019-05-27 07:14:29.000000000","submitted":"2019-05-27 07:14:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"661378-1558941269189-d19f6385","meta_rev_id":"53fc937c5d75ada7eef7ec472096cd59a153911b","_number":661378,"virtual_id_number":661378,"owner":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-05-27 07:14:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"recommended":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"all":[{"value":1,"date":"2019-05-26 09:30:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":2,"date":"2019-05-27 05:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":2,"date":"2019-05-26 06:35:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"all":[{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":1,"date":"2019-05-27 05:32:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"value":0,"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-26 06:35:05.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-05-26 09:30:59.000000000","updated_by":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"reviewer":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"state":"REVIEWER"},{"updated":"2019-05-27 05:32:38.000000000","updated_by":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"reviewer":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"state":"REVIEWER"},{"updated":"2019-05-27 07:14:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a2d58f27cd2e4698ade05864d0f9092d7c49f94","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2019-05-24 19:00:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"a17e54c918a4ce7a7341287a251cf2c1e30cf745","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-24 20:40:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/661378/1/check/openstack-ansible-linters/5f37b7a/ : SUCCESS in 4m 23s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/78/661378/1/check/openstack-ansible-deploy-aio_metal-debian-stable/7744630/ : SUCCESS in 55m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/78/661378/1/check/openstack-ansible-deploy-aio_metal-centos-7/ee96825/ : SUCCESS in 1h 22m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/78/661378/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d03b006/ : SUCCESS in 53m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/78/661378/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/f49c7ec/ : FAILURE in 40m 18s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/78/661378/1/check/openstack-tox-docs/5ecde85/html/ : SUCCESS in 2m 45s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/78/661378/1/check/openstack-ansible-uw_apache-centos-7/02370ed/ : SUCCESS in 36m 58s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/78/661378/1/check/openstack-ansible-uw_apache-ubuntu-bionic/972f5e7/ : SUCCESS in 33m 51s","accounts_in_message":[],"_revision_number":1},{"id":"afc2b5d2975ca0f53234675095d1006688058129","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-05-26 06:35:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8bf290b9d35349010f209a9020327cb8f204dd00","author":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"date":"2019-05-26 09:30:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"1da1e9fd4888a48ac585b4ea65d4c78ef1a6692e","author":{"_account_id":23509,"name":"dos","email":"fansp@awcloud.com","username":"fansp"},"date":"2019-05-26 09:31:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4b98a9daf189a2fbfd749422010001233d1065af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-26 11:15:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/661378/1/check/openstack-ansible-linters/e748302/ : SUCCESS in 4m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/78/661378/1/check/openstack-ansible-deploy-aio_metal-debian-stable/8c00aef/ : SUCCESS in 57m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/78/661378/1/check/openstack-ansible-deploy-aio_metal-centos-7/f0ae06a/ : SUCCESS in 1h 14m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/78/661378/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/efb793f/ : SUCCESS in 1h 02m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/78/661378/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/695ff8d/ : FAILURE in 39m 50s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/78/661378/1/check/openstack-tox-docs/e6f71bc/html/ : SUCCESS in 2m 06s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/78/661378/1/check/openstack-ansible-uw_apache-centos-7/ba117cd/ : SUCCESS in 38m 16s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/78/661378/1/check/openstack-ansible-uw_apache-ubuntu-bionic/6186cdb/ : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":1},{"id":"45b62b8692d1671b50ae6e2549ea00c89ca63341","author":{"_account_id":7353,"name":"Kevin Carter","email":"kevin@cloudnull.com","username":"cloudnull"},"date":"2019-05-27 05:32:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9fb6ad869c1f9874ba4cdbc05fc037d24e9d2d0a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 05:32:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"23ff69af82f975c219a48bbbeef95fdd0ed2429e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 07:14:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/78/661378/1/gate/openstack-ansible-linters/86b5b54/ : SUCCESS in 5m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/78/661378/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/1db9c65/ : SUCCESS in 59m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/78/661378/1/gate/openstack-ansible-deploy-aio_metal-centos-7/3cad70e/ : SUCCESS in 1h 40m 58s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/78/661378/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/13c00ae/ : SUCCESS in 54m 22s\n- openstack-tox-docs http://logs.openstack.org/78/661378/1/gate/openstack-tox-docs/79ddc34/html/ : SUCCESS in 2m 30s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/78/661378/1/gate/openstack-ansible-uw_apache-centos-7/1d30043/ : SUCCESS in 38m 20s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/78/661378/1/gate/openstack-ansible-uw_apache-ubuntu-bionic/a12d24b/ : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":1},{"id":"be256ec03620b3268db2f0edbef84f6689547137","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-27 07:14:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7ab6f81bbb6b2d37e1d9e2fcee01198f1f8ec071","revisions":{"7ab6f81bbb6b2d37e1d9e2fcee01198f1f8ec071":{"kind":"REWORK","_number":1,"created":"2019-05-24 19:00:38.000000000","uploader":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"ref":"refs/changes/78/661378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/78/661378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/78/661378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/78/661378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/78/661378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/78/661378/1"}}},"commit":{"parents":[{"commit":"37db9b4f157370c7f50292ab4c5c4ec9c4140907","subject":"Merge \"Remove keystone directory on source installs\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/37db9b4f157370c7f50292ab4c5c4ec9c4140907"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-05-21 16:52:02.000000000","tz":180},"committer":{"name":"Marc Gariépy (mgariepy)","email":"gariepy.marc@gmail.com","date":"2019-05-24 19:00:38.000000000","tz":0},"subject":"Update uw_apache to run against bionic","message":"Update uw_apache to run against bionic\n\nuw_apache test was runned against xenial, which is not currrently\nsupported, due to which job was updated to run on bionic nodeset.\n\nWe also need to enable proxy_uwsgi for debian based distros.\n\nCo-Authored-By: Guilherme Steinmuller Pimentel \u003cgsteinmuller@vexxhost.com\u003e\nCo-Authored-By: Marc Gariépy \u003cgariepy.marc@gmail.com\u003e\nChange-Id: Ibff3aa2a1ac1bbf2493aaf2419ee1e4dd763934c\n(cherry picked from commit d6d3bda11317c443a612344a9e5651e8dd26275d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7ab6f81bbb6b2d37e1d9e2fcee01198f1f8ec071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7ab6f81bbb6b2d37e1d9e2fcee01198f1f8ec071"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
