)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~662571","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Idbef95bc115755994156ab0fee7538370392e67d","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"ubuntu-distro-install","hashtags":[],"change_id":"Idbef95bc115755994156ab0fee7538370392e67d","subject":"Fix distro installs on Ubuntu","status":"MERGED","created":"2019-05-31 20:14:09.000000000","updated":"2019-06-11 20:30:28.000000000","submitted":"2019-06-11 08:48:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"662571-1560242912024-b5d80b2d","meta_rev_id":"cf5043838fabd7a0ff67ab739e7b2dc2d0f6aa91","_number":662571,"virtual_id_number":662571,"owner":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-05 20:50:30.000000000","_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"date":"2019-06-05 20:19:51.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-06-11 08:48:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":2,"date":"2019-06-05 20:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-11 07:04:46.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 07:04:46.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-05 20:08:06.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-05 20:19:51.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-06-11 07:04:46.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-06-11 08:48:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c52fda459adc160f517e589cd8464a134258c25b","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-05-31 20:14:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b609f1bfdfae459aaf89a7566384143096a6102","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-05-31 20:16:26.000000000","message":"Patch Set 1:\n\ncheck experimental","accounts_in_message":[],"_revision_number":1},{"id":"28a87aec892a02d03bc71ad30e124aad1267b628","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-05-31 20:21:54.000000000","message":"Topic set to ubuntu-distro-install","accounts_in_message":[],"_revision_number":1},{"id":"f916316ac9ac7eec75451afb0c39bd8e6646bc86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 20:45:54.000000000","message":"Patch Set 1:\n\nBuild failed (experimental pipeline).\n\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/71/662571/1/experimental/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/6263427/ : FAILURE in 21m 09s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-150 http://logs.openstack.org/71/662571/1/experimental/openstack-ansible-deploy-aio_distro_metal-opensuse-150/83ecef1/ : FAILURE in 24m 10s","accounts_in_message":[],"_revision_number":1},{"id":"886a0d38152962af410a96882a99846e3f797cec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-31 21:56:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/71/662571/1/check/openstack-ansible-linters/02983cf/ : SUCCESS in 4m 55s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/662571/1/check/openstack-ansible-deploy-aio_metal-debian-stable/5cd7f6d/ : SUCCESS in 1h 00m 32s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/662571/1/check/openstack-ansible-deploy-aio_metal-centos-7/9aae0bb/ : SUCCESS in 1h 22m 35s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/71/662571/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/04e2b9b/ : SUCCESS in 1h 03m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/662571/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/bbb90bd/ : SUCCESS in 56m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/662571/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/57d7782/ : SUCCESS in 1h 05m 41s\n- openstack-tox-docs http://logs.openstack.org/71/662571/1/check/openstack-tox-docs/b867fe9/html/ : SUCCESS in 3m 07s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/71/662571/1/check/openstack-ansible-uw_apache-centos-7/f532b1a/ : SUCCESS in 42m 02s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/71/662571/1/check/openstack-ansible-uw_apache-ubuntu-bionic/2b9674e/ : SUCCESS in 29m 13s","accounts_in_message":[],"_revision_number":1},{"id":"bb70e5dc1a3d6915d868c4a22c7912977e408848","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-03 15:03:11.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"9d03cb8f7e5ebbf0709c28c3276df727a8d7917e","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-03 17:46:42.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"4e220eb8c753bb8af1b26f9bf92caf59fe7fea3f","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-03 20:07:17.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"98d12947f33039f14e8256799f53ac17eae787a5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-03 21:35:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/71/662571/2/check/openstack-ansible-linters/4f4d40c/ : SUCCESS in 5m 00s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/662571/2/check/openstack-ansible-deploy-aio_metal-debian-stable/b8bc518/ : SUCCESS in 1h 04m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/662571/2/check/openstack-ansible-deploy-aio_metal-centos-7/5ac976b/ : SUCCESS in 1h 20m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/71/662571/2/check/openstack-ansible-deploy-aio_metal-opensuse-150/5c9ee48/ : SUCCESS in 1h 11m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/662571/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5c9d614/ : SUCCESS in 58m 14s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/662571/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/2944621/ : SUCCESS in 1h 02m 32s\n- openstack-tox-docs http://logs.openstack.org/71/662571/2/check/openstack-tox-docs/ec65df4/html/ : SUCCESS in 2m 51s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/71/662571/2/check/openstack-ansible-uw_apache-centos-7/be77a84/ : SUCCESS in 44m 24s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/71/662571/2/check/openstack-ansible-uw_apache-ubuntu-bionic/fe0119d/ : SUCCESS in 29m 10s","accounts_in_message":[],"_revision_number":2},{"id":"3d0b41bdac2b4b2acbb972e289bbbf3df377a0a0","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-05 20:08:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"8fd611332ae6d0445154e92389570d280ba6a1ab","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-06-05 20:19:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e7fb7fa7d77c1730a06f7286fa6a670ad676c564","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-05 20:50:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"72c40e8b445d0e626072ea01e3e9d4865cd54375","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-06-11 07:04:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"26b65520e9d5f6f2b5a38cfeb92814feada4461a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 07:04:58.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ce1ca46c66c06e2d1ce30228a34800c17d851cde","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 08:48:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/71/662571/2/gate/openstack-ansible-linters/978a492/ : SUCCESS in 9m 38s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/71/662571/2/gate/openstack-ansible-deploy-aio_metal-debian-stable/c3a0114/ : SUCCESS in 1h 04m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/71/662571/2/gate/openstack-ansible-deploy-aio_metal-centos-7/b50f2d3/ : SUCCESS in 1h 35m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/71/662571/2/gate/openstack-ansible-deploy-aio_metal-opensuse-150/178f423/ : SUCCESS in 57m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/71/662571/2/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/40fd3c4/ : SUCCESS in 59m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/71/662571/2/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/f9a5ace/ : SUCCESS in 1h 08m 49s\n- openstack-tox-docs http://logs.openstack.org/71/662571/2/gate/openstack-tox-docs/89df43f/html/ : SUCCESS in 3m 12s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/71/662571/2/gate/openstack-ansible-uw_apache-centos-7/4adb10e/ : SUCCESS in 37m 15s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/71/662571/2/gate/openstack-ansible-uw_apache-ubuntu-bionic/e961ba4/ : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":2},{"id":"3ef753e229e135dae0ab4b8cccf04cf52293ef8b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 08:48:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"cf5043838fabd7a0ff67ab739e7b2dc2d0f6aa91","author":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"date":"2019-06-11 20:30:28.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit b4e2f058e1b93a24e7a74a4ffb8e6c13621c8627","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"254a447ffeffff5dca772fd9d1879ee5c8a0ac66","revisions":{"d21608f2b0629e9c19d28170888194d48e6699ea":{"kind":"REWORK","_number":1,"created":"2019-05-31 20:14:09.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/71/662571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/71/662571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/662571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/662571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/662571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/662571/1"}}},"commit":{"parents":[{"commit":"4d88d04f4480e9a1a95ac199d10886642ebac11b","subject":"Merge \"Clarify supported SUSE versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/4d88d04f4480e9a1a95ac199d10886642ebac11b"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-05-31 20:02:17.000000000","tz":-420},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-05-31 20:13:04.000000000","tz":-420},"subject":"Fix distro installs on Ubuntu","message":"Fix distro installs on Ubuntu\n\nBeginning in the Stein release, Ubuntu distro packages are now using\nPython3. This requires additionally installing and using the uwsgi python3\nplugin.\n\nChange-Id: Idbef95bc115755994156ab0fee7538370392e67d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d21608f2b0629e9c19d28170888194d48e6699ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d21608f2b0629e9c19d28170888194d48e6699ea"}]},"branch":"refs/heads/master"},"254a447ffeffff5dca772fd9d1879ee5c8a0ac66":{"kind":"REWORK","_number":2,"created":"2019-06-03 20:07:17.000000000","uploader":{"_account_id":14805,"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","username":"jimmy-mccrory"},"ref":"refs/changes/71/662571/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/71/662571/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/662571/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/662571/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/662571/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/662571/2"}}},"commit":{"parents":[{"commit":"4d88d04f4480e9a1a95ac199d10886642ebac11b","subject":"Merge \"Clarify supported SUSE versions\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/4d88d04f4480e9a1a95ac199d10886642ebac11b"}]}],"author":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-05-31 20:02:17.000000000","tz":-420},"committer":{"name":"Jimmy McCrory","email":"jimmy.mccrory@gmail.com","date":"2019-06-03 20:05:41.000000000","tz":-420},"subject":"Fix distro installs on Ubuntu","message":"Fix distro installs on Ubuntu\n\nBeginning in the Stein release, Ubuntu distro packages are now using\nPython3. This requires additionally installing and using the uwsgi python3\nplugin.\n\nThe keystone package includes a dependency for apache2, python3-keystone\nshould be used instead.\n\nChange-Id: Idbef95bc115755994156ab0fee7538370392e67d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/254a447ffeffff5dca772fd9d1879ee5c8a0ac66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/254a447ffeffff5dca772fd9d1879ee5c8a0ac66"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
