)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~664093","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fqueens~Ib233b3837f1266e46efe76d25409236f988c6a13","project":"openstack/openstack-ansible-os_keystone","branch":"stable/queens","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ib233b3837f1266e46efe76d25409236f988c6a13","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-07 23:25:37.000000000","updated":"2019-06-11 21:58:49.000000000","submitted":"2019-06-11 21:58:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664093-1560290329797-8e310be5","meta_rev_id":"f93a7088fd2f305e055c593d04b3b516a4bf0edc","_number":664093,"virtual_id_number":664093,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-11 21:58:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-11 20:56:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-11 20:56:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-11 20:56:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-11 21:58:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c8f72c787667718dc8801f7ff878b720b2ab0e1b","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-07 23:25:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b65f91f0470101a6142dd7c4f84634ea63bd8a98","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-08 00:53:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs http://logs.openstack.org/93/664093/1/check/build-openstack-sphinx-docs/1815b8e/html/ : SUCCESS in 3m 43s\n- openstack-ansible-linters http://logs.openstack.org/93/664093/1/check/openstack-ansible-linters/0a7d112/ : SUCCESS in 4m 58s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/93/664093/1/check/openstack-ansible-functional-centos-7/2b2c41c/ : SUCCESS in 46m 50s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/93/664093/1/check/openstack-ansible-functional-opensuse-423/194320d/ : FAILURE in 9m 04s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/93/664093/1/check/openstack-ansible-functional-ubuntu-xenial/91c7820/ : SUCCESS in 28m 46s\n- openstack-ansible-uw_apache-centos-7-nv http://logs.openstack.org/93/664093/1/check/openstack-ansible-uw_apache-centos-7-nv/63cf25f/ : SUCCESS in 55m 23s (non-voting)\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/93/664093/1/check/openstack-ansible-uw_apache-ubuntu-xenial/609fab9/ : SUCCESS in 37m 01s","accounts_in_message":[],"_revision_number":1},{"id":"61f2a27298a16ed8157904d56694696b8d4c2267","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 19:31:27.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1707eac85c64bd7aefc9db779b56a8debf0149d7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 20:33:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/93/664093/1/check/build-openstack-sphinx-docs/1519183/html/ : SUCCESS in 3m 55s\n- openstack-ansible-linters http://logs.openstack.org/93/664093/1/check/openstack-ansible-linters/72ad1a5/ : SUCCESS in 6m 39s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/93/664093/1/check/openstack-ansible-functional-centos-7/5fbd224/ : SUCCESS in 55m 14s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/93/664093/1/check/openstack-ansible-functional-ubuntu-xenial/3c47277/ : SUCCESS in 29m 58s\n- openstack-ansible-uw_apache-centos-7-nv http://logs.openstack.org/93/664093/1/check/openstack-ansible-uw_apache-centos-7-nv/ada4e7b/ : SUCCESS in 54m 44s (non-voting)\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/93/664093/1/check/openstack-ansible-uw_apache-ubuntu-xenial/044782b/ : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":1},{"id":"cae2986ee2323972e4d3c11b27a0f68992441bbc","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-11 20:56:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6cc8e8e48f5df8d3ad212feb093393fd1885097c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 20:56:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0f257ba8635bf0093a96f47cd0e8c694d77cc3fc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:58:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/93/664093/1/gate/build-openstack-sphinx-docs/9cc5a15/html/ : SUCCESS in 3m 22s\n- openstack-ansible-linters http://logs.openstack.org/93/664093/1/gate/openstack-ansible-linters/8fd0d1e/ : SUCCESS in 6m 03s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/93/664093/1/gate/openstack-ansible-functional-centos-7/afa303e/ : SUCCESS in 48m 43s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/93/664093/1/gate/openstack-ansible-functional-ubuntu-xenial/f436a72/ : SUCCESS in 32m 44s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/93/664093/1/gate/openstack-ansible-uw_apache-ubuntu-xenial/d3f66f9/ : SUCCESS in 32m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f647f2c16e3d46f1e27d3ebdbb086247a09b67e1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 21:58:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"30d7f6536cce7c6a150c8314df59223823741414","revisions":{"30d7f6536cce7c6a150c8314df59223823741414":{"kind":"REWORK","_number":1,"created":"2019-06-07 23:25:37.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/664093/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/93/664093/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/664093/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/664093/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/664093/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/664093/1"}}},"commit":{"parents":[{"commit":"17db6940071361c40c3740e39a9e1580cb45dc5a","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/17db6940071361c40c3740e39a9e1580cb45dc5a"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:25:32.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-07 23:25:32.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ib233b3837f1266e46efe76d25409236f988c6a13\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/30d7f6536cce7c6a150c8314df59223823741414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/30d7f6536cce7c6a150c8314df59223823741414"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
