)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~664323","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Frocky~Iad8ae22ff80bae72e854f7cd255bec0b6c39bde8","project":"openstack/openstack-ansible-os_keystone","branch":"stable/rocky","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Iad8ae22ff80bae72e854f7cd255bec0b6c39bde8","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-10 15:44:30.000000000","updated":"2019-06-20 22:22:32.000000000","submitted":"2019-06-20 22:22:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664323-1561069352817-5bb55d27","meta_rev_id":"6ba0ba7afcdabcea5bd51f78dc6ed25cf3d48a4e","_number":664323,"virtual_id_number":664323,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-20 22:22:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-20 20:07:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-20 21:24:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-20 21:24:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 20:07:43.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-20 21:24:57.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-20 22:22:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c24386c2a86f0557a82af2c917eb29459d3e427a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-10 15:44:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d55e17dfe114cb443ca5fb0f226bccf30e59c0e8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-10 17:02:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/23/664323/1/check/openstack-ansible-linters/38aa8ca/ : SUCCESS in 5m 37s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/23/664323/1/check/openstack-ansible-functional-centos-7/434a27a/ : SUCCESS in 38m 08s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/23/664323/1/check/openstack-ansible-functional-opensuse-423/c6006aa/ : FAILURE in 8m 08s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/23/664323/1/check/openstack-ansible-functional-opensuse-150/4ea1446/ : FAILURE in 9m 42s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/23/664323/1/check/openstack-ansible-functional-ubuntu-bionic/ef58836/ : SUCCESS in 27m 52s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/23/664323/1/check/openstack-ansible-functional-ubuntu-xenial/6427d71/ : SUCCESS in 33m 30s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/664323/1/check/build-openstack-sphinx-docs/4cf6dbd/html/ : SUCCESS in 3m 29s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/23/664323/1/check/openstack-ansible-uw_apache-centos-7/25e95d6/ : FAILURE in 16m 49s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/23/664323/1/check/openstack-ansible-uw_apache-ubuntu-xenial/afab71e/ : SUCCESS in 33m 45s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/23/664323/1/check/openstack-ansible-functional-distro_install-ubuntu-xenial/9f8306b/ : SUCCESS in 31m 29s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/23/664323/1/check/openstack-ansible-functional-distro_install-centos-7/82dc5bb/ : SUCCESS in 42m 06s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/23/664323/1/check/openstack-ansible-functional-distro_install-opensuse-423/0a0fa15/ : FAILURE in 8m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"515d73a612b59d16e5460257437e923fc39f2186","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-20 20:01:12.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3f60924c28b0f94f80ff4811378f2a5153f098f9","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-20 20:07:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"b47c005f0a159e8901ac664faa630d6a80e075bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:23:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/23/664323/2/check/openstack-ansible-linters/1cfa84e/ : SUCCESS in 6m 17s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/23/664323/2/check/openstack-ansible-functional-centos-7/be86ee3/ : SUCCESS in 38m 39s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/23/664323/2/check/openstack-ansible-functional-opensuse-423/ffd4f74/ : FAILURE in 8m 06s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/23/664323/2/check/openstack-ansible-functional-opensuse-150/0179672/ : FAILURE in 8m 37s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/23/664323/2/check/openstack-ansible-functional-ubuntu-bionic/eeb9bb4/ : SUCCESS in 30m 38s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/23/664323/2/check/openstack-ansible-functional-ubuntu-xenial/c6bcb8a/ : SUCCESS in 28m 29s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/664323/2/check/build-openstack-sphinx-docs/6661920/html/ : SUCCESS in 4m 34s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/23/664323/2/check/openstack-ansible-uw_apache-centos-7/e153c74/ : SUCCESS in 37m 01s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/23/664323/2/check/openstack-ansible-uw_apache-ubuntu-xenial/8736f1f/ : SUCCESS in 33m 45s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/23/664323/2/check/openstack-ansible-functional-distro_install-ubuntu-xenial/be2f3c8/ : SUCCESS in 1h 11m 06s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/23/664323/2/check/openstack-ansible-functional-distro_install-centos-7/f9dfe4f/ : SUCCESS in 31m 56s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/23/664323/2/check/openstack-ansible-functional-distro_install-opensuse-423/8fa2dba/ : FAILURE in 7m 17s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"7bf34c4890348a3cae02c065135ead95ce93cc01","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-20 21:24:48.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"bda1a5b6fafb7b5ef48db54f10eebded196c5651","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-20 21:24:57.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"1d28b64284694f00d288c36a4649fa7d61eaedad","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 21:25:09.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9845b36de59aa7445b9291046d5fcdc2601d054a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:22:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b95b1ede72a772c574e0040f327878674b287a4e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-20 22:22:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/23/664323/2/gate/openstack-ansible-linters/88dfff0/ : SUCCESS in 5m 44s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/23/664323/2/gate/openstack-ansible-functional-centos-7/7bf9ff1/ : SUCCESS in 31m 52s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/23/664323/2/gate/openstack-ansible-functional-opensuse-423/1100d9e/ : FAILURE in 10m 09s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/23/664323/2/gate/openstack-ansible-functional-ubuntu-bionic/065940a/ : SUCCESS in 30m 07s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/23/664323/2/gate/openstack-ansible-functional-ubuntu-xenial/4c1b006/ : SUCCESS in 29m 47s\n- build-openstack-sphinx-docs http://logs.openstack.org/23/664323/2/gate/build-openstack-sphinx-docs/c0eb67c/html/ : SUCCESS in 3m 23s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/23/664323/2/gate/openstack-ansible-uw_apache-centos-7/82a6978/ : SUCCESS in 42m 57s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/23/664323/2/gate/openstack-ansible-uw_apache-ubuntu-xenial/5915eeb/ : SUCCESS in 37m 45s\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/23/664323/2/gate/openstack-ansible-functional-distro_install-ubuntu-bionic/9e206a0/ : SUCCESS in 39m 59s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/23/664323/2/gate/openstack-ansible-functional-distro_install-centos-7/caf1fbe/ : SUCCESS in 32m 58s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/23/664323/2/gate/openstack-ansible-functional-distro_install-opensuse-423/17d8637/ : FAILURE in 9m 54s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1850b420d35b560ee35dab2add587b563cf8adc2","revisions":{"f245ca1c08db555ffaa0a4597c8b92e387bd4dd4":{"kind":"REWORK","_number":1,"created":"2019-06-10 15:44:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/664323/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/23/664323/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/23/664323/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/23/664323/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/23/664323/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/23/664323/1"}}},"commit":{"parents":[{"commit":"bbe39f9fc484b284eb520ba7bbdd4bd407aef589","subject":"Merge \"Remove unsupported option for keystone mapping\" into stable/rocky","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/bbe39f9fc484b284eb520ba7bbdd4bd407aef589"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:29.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-10 15:44:29.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iad8ae22ff80bae72e854f7cd255bec0b6c39bde8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f245ca1c08db555ffaa0a4597c8b92e387bd4dd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f245ca1c08db555ffaa0a4597c8b92e387bd4dd4"}]},"branch":"refs/heads/stable/rocky"},"1850b420d35b560ee35dab2add587b563cf8adc2":{"kind":"REWORK","_number":2,"created":"2019-06-20 20:01:12.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/23/664323/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/23/664323/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/23/664323/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/23/664323/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/23/664323/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/23/664323/2"}}},"commit":{"parents":[{"commit":"8703d6a232b3a017a5f0f6e74aa4cd8c0d09b1d5","subject":"Add headers to allow the inline script for SSO redirect.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8703d6a232b3a017a5f0f6e74aa4cd8c0d09b1d5"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:01:10.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-20 20:01:10.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Iad8ae22ff80bae72e854f7cd255bec0b6c39bde8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1850b420d35b560ee35dab2add587b563cf8adc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1850b420d35b560ee35dab2add587b563cf8adc2"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
