)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~664724","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ie9a17b13c825d3406e65dcace3ad4806d49fc871","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ie9a17b13c825d3406e65dcace3ad4806d49fc871","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-11 22:18:10.000000000","updated":"2019-06-12 07:55:39.000000000","submitted":"2019-06-12 07:55:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664724-1560326139541-48a02ede","meta_rev_id":"e8d12388c0122c72d0fef958cb86ad6400b8538f","_number":664724,"virtual_id_number":664724,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-06-12 07:55:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-12 06:14:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-12 06:14:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 06:14:52.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-12 07:55:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4ad4abbb90429e0126a253bf7e700ce81e33f5fe","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-11 22:18:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b956bf91ba579e2ea4bc8fc69cd92d3b1ceb977","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-11 23:54:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/24/664724/1/check/openstack-ansible-linters/5a35e46/ : SUCCESS in 5m 07s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/24/664724/1/check/openstack-ansible-deploy-aio_metal-debian-stable/20da4a1/ : SUCCESS in 1h 04m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/24/664724/1/check/openstack-ansible-deploy-aio_metal-centos-7/e0114cd/ : SUCCESS in 1h 26m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/24/664724/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/56c698c/ : SUCCESS in 1h 02m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/24/664724/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3ce365c/ : SUCCESS in 1h 09m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/24/664724/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/9bd1d3b/ : SUCCESS in 1h 08m 58s\n- openstack-tox-docs http://logs.openstack.org/24/664724/1/check/openstack-tox-docs/f42911a/html/ : SUCCESS in 3m 36s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/24/664724/1/check/openstack-ansible-uw_apache-centos-7/03e9e5e/ : SUCCESS in 42m 03s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/24/664724/1/check/openstack-ansible-uw_apache-ubuntu-bionic/d9af0ab/ : SUCCESS in 29m 44s","accounts_in_message":[],"_revision_number":1},{"id":"1cf9959ad293307238ba83135ed10144382d2661","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-12 06:14:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5cef1b27bbbe6beab2263828ddfc742caf549308","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 06:15:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3fce633f88317ce98c75b254b6ac2166b33b844","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 07:55:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"403027f630b13328d1c5688018f473a72f85d09e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 07:55:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/24/664724/1/gate/openstack-ansible-linters/7ed13b5/ : SUCCESS in 5m 33s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/24/664724/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/587afc3/ : SUCCESS in 1h 00m 29s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/24/664724/1/gate/openstack-ansible-deploy-aio_metal-centos-7/6c19c84/ : SUCCESS in 1h 20m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/24/664724/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/e162f17/ : SUCCESS in 1h 10m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/24/664724/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/dbcb96a/ : SUCCESS in 55m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/24/664724/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/6ca27fc/ : SUCCESS in 56m 45s\n- openstack-tox-docs http://logs.openstack.org/24/664724/1/gate/openstack-tox-docs/4dfc3d2/html/ : SUCCESS in 2m 42s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/24/664724/1/gate/openstack-ansible-uw_apache-centos-7/c7f7f38/ : SUCCESS in 37m 03s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/24/664724/1/gate/openstack-ansible-uw_apache-ubuntu-bionic/9981bd8/ : SUCCESS in 29m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"04ac5b4243111b81135108461a8ab099c1c3ae37","revisions":{"04ac5b4243111b81135108461a8ab099c1c3ae37":{"kind":"REWORK","_number":1,"created":"2019-06-11 22:18:10.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/24/664724/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/24/664724/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/664724/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/664724/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/664724/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/24/664724/1"}}},"commit":{"parents":[{"commit":"a884451b5b786d3f054d33e4d1589cacd2e170af","subject":"Merge \"Execute the keystone db setup earlier in the process\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a884451b5b786d3f054d33e4d1589cacd2e170af"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:18:08.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-11 22:18:08.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ie9a17b13c825d3406e65dcace3ad4806d49fc871\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/04ac5b4243111b81135108461a8ab099c1c3ae37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/04ac5b4243111b81135108461a8ab099c1c3ae37"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
