)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~664753","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fstein~Ifec88251712a99ad7afbb51e386adf15bc4841eb","project":"openstack/openstack-ansible-os_keystone","branch":"stable/stein","topic":"bug/1830397","hashtags":[],"change_id":"Ifec88251712a99ad7afbb51e386adf15bc4841eb","subject":"Prevent autoremoval of packages on bare metal","status":"MERGED","created":"2019-06-12 00:41:40.000000000","updated":"2019-07-01 16:25:08.000000000","submitted":"2019-07-01 16:25:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"664753-1561998308175-427e186e","meta_rev_id":"5dd23ddd60528711e91ad530d5469608b138e0c6","_number":664753,"virtual_id_number":664753,"owner":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":2,"date":"2019-07-01 16:25:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"all":[{"value":2,"date":"2019-07-01 15:01:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-12 11:04:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":1,"date":"2019-06-12 01:58:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},{"value":2,"date":"2019-06-12 07:08:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"all":[{"value":1,"date":"2019-07-01 15:01:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},{"value":0,"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-12 01:58:07.000000000","updated_by":{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},"reviewer":{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},"state":"REVIEWER"},{"updated":"2019-06-12 07:08:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-12 11:04:54.000000000","updated_by":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"reviewer":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"state":"REVIEWER"},{"updated":"2019-07-01 15:01:12.000000000","updated_by":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"reviewer":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"state":"REVIEWER"},{"updated":"2019-07-01 16:25:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e3baa07312316f109e31b8f82b2e0ea37bb9a047","author":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"date":"2019-06-12 00:41:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"915d4767c7b34a5c33c354b4bfa56e5a197e32dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-12 01:56:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/53/664753/1/check/openstack-ansible-linters/cf5768b/ : SUCCESS in 5m 13s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/53/664753/1/check/openstack-ansible-deploy-aio_metal-debian-stable/1ca7d87/ : SUCCESS in 46m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/53/664753/1/check/openstack-ansible-deploy-aio_metal-centos-7/62ef42b/ : SUCCESS in 1h 05m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/53/664753/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/fb31141/ : SUCCESS in 54m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/53/664753/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/cc9802c/ : FAILURE in 38m 54s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/53/664753/1/check/openstack-tox-docs/264021d/html/ : SUCCESS in 2m 54s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/53/664753/1/check/openstack-ansible-uw_apache-centos-7/1c4584c/ : SUCCESS in 35m 36s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/53/664753/1/check/openstack-ansible-uw_apache-ubuntu-bionic/f7b1dd4/ : SUCCESS in 36m 58s","accounts_in_message":[],"_revision_number":1},{"id":"464becacc3a7c4de1c866750b881cd64f5ac5a52","author":{"_account_id":30127,"name":"John Hou","email":"houj@awcloud.com","username":"houj"},"date":"2019-06-12 01:58:07.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c771520ed55078c5e468af7d1edee1d80a850248","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-12 07:08:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"61ba62822425f473e12e7a5de06938d178a23ecc","author":{"_account_id":30381,"name":"Jean-Pierre REN","email":"renzp@awcloud.com","username":"jpren"},"date":"2019-06-12 11:04:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"e4bc7591ad446c0bf9c3edebc65a8b92c9b5dc17","author":{"_account_id":15993,"name":"Amy Marrich","display_name":"Amy Marrich (spotz)","email":"amy@demarco.com","username":"amarrich"},"date":"2019-07-01 15:01:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"67ff33e0192f00d621bc1a4c505b003151a733d5","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 15:01:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62c748838e7d6dde54938929ccb2b583eb1676eb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 16:25:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/53/664753/1/gate/openstack-ansible-linters/2189143/ : SUCCESS in 4m 54s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/53/664753/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/e66b671/ : SUCCESS in 58m 08s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/53/664753/1/gate/openstack-ansible-deploy-aio_metal-centos-7/7e2a3cc/ : SUCCESS in 1h 16m 25s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/53/664753/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/56938e9/ : SUCCESS in 58m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/53/664753/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/5b68407/ : SUCCESS in 55m 24s\n- openstack-tox-docs http://logs.openstack.org/53/664753/1/gate/openstack-tox-docs/d43bd07/html/ : SUCCESS in 2m 47s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/53/664753/1/gate/openstack-ansible-uw_apache-centos-7/c72f363/ : SUCCESS in 37m 01s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/53/664753/1/gate/openstack-ansible-uw_apache-ubuntu-bionic/bf623d6/ : SUCCESS in 29m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a5284631282b0fbca0d0eca913b4c529cf8927ed","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-01 16:25:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aee98634496f457177652307a9f4613cdd290c36","revisions":{"aee98634496f457177652307a9f4613cdd290c36":{"kind":"REWORK","_number":1,"created":"2019-06-12 00:41:40.000000000","uploader":{"_account_id":14552,"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","username":"bjoern"},"ref":"refs/changes/53/664753/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/53/664753/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/53/664753/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/53/664753/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/53/664753/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/53/664753/1"}}},"commit":{"parents":[{"commit":"5332d9609e1007c405ec33efd7cf91c0f3161079","subject":"Merge \"Fix distro installs on Ubuntu\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5332d9609e1007c405ec33efd7cf91c0f3161079"}]}],"author":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-05-24 15:24:36.000000000","tz":-300},"committer":{"name":"Bjoern Teipel","email":"bjoern.teipel@rackspace.com","date":"2019-06-12 00:41:31.000000000","tz":-300},"subject":"Prevent autoremoval of packages on bare metal","message":"Prevent autoremoval of packages on bare metal\n\nWhen deploying OSA on bare metal (no container tech),\nthe os keystone role interferes with os horizon as the\npackage apache2 is automatically uninstalled.\n\nChange-Id: Ifec88251712a99ad7afbb51e386adf15bc4841eb\nCloses-Bug: #1830397\n(cherry picked from commit ae6c1982ec2101547b6744cf157aae4ce5bbb49e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/aee98634496f457177652307a9f4613cdd290c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/aee98634496f457177652307a9f4613cdd290c36"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
