)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~665500","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Frocky~I7f6b80acb27fb959ce6d68bd35d3cc5dccc86f9f","project":"openstack/openstack-ansible-os_keystone","branch":"stable/rocky","topic":"policy-in-code","hashtags":[],"change_id":"I7f6b80acb27fb959ce6d68bd35d3cc5dccc86f9f","subject":"Fix custom policy configuration","status":"MERGED","created":"2019-06-15 02:00:59.000000000","updated":"2019-07-10 16:50:46.000000000","submitted":"2019-07-10 16:50:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"665500-1562777446655-dffa1978","meta_rev_id":"882c353874cd737b0346efc00a20f6f617a782f6","_number":665500,"virtual_id_number":665500,"owner":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2019-06-15 17:39:56.000000000","_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-10 16:50:46.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-07-10 16:02:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-20 20:16:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":1,"date":"2019-07-10 16:02:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-20 20:16:59.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-10 16:02:36.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-10 16:50:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2568ebbc4df7f43e186ee66b9241155f9925b55b","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-15 02:00:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"05514919a397303cec3b96d4883ef83e2f0e23ca","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 03:51:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/00/665500/1/check/openstack-ansible-linters/545a7df/ : SUCCESS in 4m 52s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-centos-7/bf052c3/ : SUCCESS in 37m 36s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-opensuse-423/a845efd/ : FAILURE in 9m 13s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-opensuse-150/5dc302d/ : RETRY_LIMIT in 2m 40s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-ubuntu-bionic/87e7391/ : TIMED_OUT in 1h 31m 00s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-ubuntu-xenial/c89d140/ : SUCCESS in 30m 33s\n- build-openstack-sphinx-docs http://logs.openstack.org/00/665500/1/check/build-openstack-sphinx-docs/30b31ad/html/ : SUCCESS in 3m 26s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/00/665500/1/check/openstack-ansible-uw_apache-centos-7/a81574e/ : SUCCESS in 39m 00s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/00/665500/1/check/openstack-ansible-uw_apache-ubuntu-xenial/7f0908e/ : SUCCESS in 31m 09s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-distro_install-ubuntu-xenial/fb03638/ : SUCCESS in 28m 51s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-distro_install-centos-7/4063f6e/ : SUCCESS in 34m 32s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-distro_install-opensuse-423/3f9142c/ : FAILURE in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"af401cee8411970149cf97e6bf4b3ff66e7dc92b","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-15 17:39:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6e4b516f9af57c173e40f53f33376fd016fce5f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-15 18:26:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/00/665500/1/check/openstack-ansible-linters/5bde7b0/ : SUCCESS in 5m 46s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-centos-7/8b10407/ : SUCCESS in 40m 38s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-opensuse-423/e43476c/ : FAILURE in 8m 54s (non-voting)\n- openstack-ansible-functional-opensuse-150 http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-opensuse-150/08ebb22/ : RETRY_LIMIT in 4m 19s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-ubuntu-bionic/9ac291a/ : SUCCESS in 32m 27s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-ubuntu-xenial/5f8c829/ : SUCCESS in 43m 31s\n- build-openstack-sphinx-docs http://logs.openstack.org/00/665500/1/check/build-openstack-sphinx-docs/8c1812c/html/ : SUCCESS in 2m 58s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/00/665500/1/check/openstack-ansible-uw_apache-centos-7/d756a38/ : SUCCESS in 41m 01s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/00/665500/1/check/openstack-ansible-uw_apache-ubuntu-xenial/0dfad1a/ : SUCCESS in 37m 18s\n- openstack-ansible-functional-distro_install-ubuntu-xenial http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-distro_install-ubuntu-xenial/5e82480/ : SUCCESS in 31m 04s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-distro_install-centos-7/d4e1d0f/ : SUCCESS in 41m 09s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/00/665500/1/check/openstack-ansible-functional-distro_install-opensuse-423/a674a73/ : FAILURE in 7m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"608cf6d370d5cfb18d8dfa33ad939a9e6c168890","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-20 20:16:59.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"554a7b5ced1a5ccdf6a6cbfbb908e80ff8bf483c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-10 16:02:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1bfff23507e75b1f2bf5b6bf227be1f4d8ba843d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:02:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab5b37386e1ff6c90dbfba41ecfc9da576cfb5da","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:50:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c6cd3276a84bb81c267760e6a3a2cfdb35ea8f5e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 16:50:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/00/665500/1/gate/openstack-ansible-linters/2dda959/ : SUCCESS in 5m 44s\n- openstack-ansible-functional-centos-7 http://logs.openstack.org/00/665500/1/gate/openstack-ansible-functional-centos-7/c2c2a46/ : SUCCESS in 34m 06s\n- openstack-ansible-functional-opensuse-423 http://logs.openstack.org/00/665500/1/gate/openstack-ansible-functional-opensuse-423/297f3de/ : FAILURE in 9m 17s (non-voting)\n- openstack-ansible-functional-ubuntu-bionic http://logs.openstack.org/00/665500/1/gate/openstack-ansible-functional-ubuntu-bionic/7be817f/ : SUCCESS in 40m 12s\n- openstack-ansible-functional-ubuntu-xenial http://logs.openstack.org/00/665500/1/gate/openstack-ansible-functional-ubuntu-xenial/26d26b4/ : SUCCESS in 37m 58s\n- build-openstack-sphinx-docs http://logs.openstack.org/00/665500/1/gate/build-openstack-sphinx-docs/00eed02/html/ : SUCCESS in 3m 57s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/00/665500/1/gate/openstack-ansible-uw_apache-centos-7/6803d39/ : SUCCESS in 38m 00s\n- openstack-ansible-uw_apache-ubuntu-xenial http://logs.openstack.org/00/665500/1/gate/openstack-ansible-uw_apache-ubuntu-xenial/0e715cd/ : SUCCESS in 28m 39s\n- openstack-ansible-functional-distro_install-ubuntu-bionic http://logs.openstack.org/00/665500/1/gate/openstack-ansible-functional-distro_install-ubuntu-bionic/be18b77/ : SUCCESS in 30m 13s\n- openstack-ansible-functional-distro_install-centos-7 http://logs.openstack.org/00/665500/1/gate/openstack-ansible-functional-distro_install-centos-7/e2e7ddf/ : SUCCESS in 33m 49s\n- openstack-ansible-functional-distro_install-opensuse-423 http://logs.openstack.org/00/665500/1/gate/openstack-ansible-functional-distro_install-opensuse-423/eb8edb9/ : FAILURE in 10m 36s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1aa7f610cc7f592d8c3d5047ce9376060990fd6e","revisions":{"1aa7f610cc7f592d8c3d5047ce9376060990fd6e":{"kind":"REWORK","_number":1,"created":"2019-06-15 02:00:59.000000000","uploader":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"ref":"refs/changes/00/665500/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/00/665500/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/665500/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/665500/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/665500/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/00/665500/1"}}},"commit":{"parents":[{"commit":"8703d6a232b3a017a5f0f6e74aa4cd8c0d09b1d5","subject":"Add headers to allow the inline script for SSO redirect.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8703d6a232b3a017a5f0f6e74aa4cd8c0d09b1d5"}]}],"author":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-06-15 01:55:13.000000000","tz":-300},"committer":{"name":"Logan V","email":"logan2211@gmail.com","date":"2019-06-15 02:00:48.000000000","tz":-300},"subject":"Fix custom policy configuration","message":"Fix custom policy configuration\n\nCustom policies for Keystone are implemented completely different\nfrom other roles, in a way that is broken from prior OSA branches.\n\nOn older and newer branches, keystone_policy_overrides is used to\nprovide a yaml dict of policy overrides using ansible vars. The\ncurrent implementation of keystone\u0027s policy distribution requires\nthat a base policy be configured in a flat file on the deployment\nhost at /etc/openstack_deploy/keystone/policy.json, to which policy\noverrides are then applied and copied to the keystone hosts.\n\nIf the flat file does not exist, no policy overrides are applied\nto Keystone.\n\nThis change modifies the policy distribution to work similarly to\nother OSA roles in the rocky branch. No flat file is sourced from\n/etc/openstack_deploy any longer, only the keystone_policy_overrides\nvar is used, and a policy.json is only deployed to Keystone\nif this var is defined by the deployer.\n\nChange-Id: I7f6b80acb27fb959ce6d68bd35d3cc5dccc86f9f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1aa7f610cc7f592d8c3d5047ce9376060990fd6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1aa7f610cc7f592d8c3d5047ce9376060990fd6e"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
