)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~666110","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ia6feeb3a386026e88b9557ecc9448f11084f2375","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"Ia6feeb3a386026e88b9557ecc9448f11084f2375","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-06-18 18:15:58.000000000","updated":"2019-06-18 21:44:22.000000000","submitted":"2019-06-18 21:44:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"666110-1560894262988-eb0806c3","meta_rev_id":"99112aedf1b84a779ffa6d50958c4172a0c35bc6","_number":666110,"virtual_id_number":666110,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":2,"date":"2019-06-18 21:44:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"all":[{"value":2,"date":"2019-06-18 18:54:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-06-18 20:10:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-06-18 20:10:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-06-18 18:54:21.000000000","updated_by":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"reviewer":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"state":"REVIEWER"},{"updated":"2019-06-18 20:10:33.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-06-18 21:44:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"44fd0a6a335c21eedf129e1cd5320b4ad8f440d9","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-06-18 18:15:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04d049438b6ac44f38b19c6482864360e1e17283","author":{"_account_id":17799,"name":"Logan V","email":"logan2211@gmail.com","username":"Logan2211"},"date":"2019-06-18 18:54:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1da244481cf1cb3d45b4bdf1e809b0e30a91a2c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 20:06:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/10/666110/1/check/openstack-ansible-linters/f498429/ : SUCCESS in 6m 03s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/10/666110/1/check/openstack-ansible-deploy-aio_metal-debian-stable/5e3724f/ : SUCCESS in 1h 13m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/10/666110/1/check/openstack-ansible-deploy-aio_metal-centos-7/19c522d/ : SUCCESS in 1h 16m 16s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/10/666110/1/check/openstack-ansible-deploy-aio_metal-opensuse-150/806f939/ : SUCCESS in 1h 11m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/10/666110/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3377599/ : SUCCESS in 1h 00m 18s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/10/666110/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/114df51/ : SUCCESS in 56m 50s\n- openstack-tox-docs http://logs.openstack.org/10/666110/1/check/openstack-tox-docs/98992ab/html/ : SUCCESS in 2m 49s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/10/666110/1/check/openstack-ansible-uw_apache-centos-7/1f634a5/ : SUCCESS in 39m 24s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/10/666110/1/check/openstack-ansible-uw_apache-ubuntu-bionic/1e7263c/ : SUCCESS in 30m 07s","accounts_in_message":[],"_revision_number":1},{"id":"f8c117ce44c6239397c60a9f7f4f65e567e96d57","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-18 20:10:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d844da3e08f3a2141da9d658b2d1b624c8e232be","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 20:10:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"60a9c0c555a250372544de7f29859a6b91291928","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:44:22.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"358a6acb63dbcfbb48476ef531439487aab51030","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-18 21:44:22.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/10/666110/1/gate/openstack-ansible-linters/e126753/ : SUCCESS in 5m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/10/666110/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/dc47404/ : SUCCESS in 1h 24m 11s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/10/666110/1/gate/openstack-ansible-deploy-aio_metal-centos-7/f4fea8c/ : SUCCESS in 1h 24m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-150 http://logs.openstack.org/10/666110/1/gate/openstack-ansible-deploy-aio_metal-opensuse-150/2b161f3/ : SUCCESS in 1h 07m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/10/666110/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3be75a1/ : SUCCESS in 1h 00m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/10/666110/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/6a29854/ : SUCCESS in 1h 00m 46s\n- openstack-tox-docs http://logs.openstack.org/10/666110/1/gate/openstack-tox-docs/7f759a7/html/ : SUCCESS in 3m 10s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/10/666110/1/gate/openstack-ansible-uw_apache-centos-7/887d27c/ : SUCCESS in 41m 15s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/10/666110/1/gate/openstack-ansible-uw_apache-ubuntu-bionic/e67a150/ : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"03b0aaf019d44e2fff3658a65c4f5cf503f61d35","revisions":{"03b0aaf019d44e2fff3658a65c4f5cf503f61d35":{"kind":"REWORK","_number":1,"created":"2019-06-18 18:15:58.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/10/666110/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/10/666110/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/10/666110/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/10/666110/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/10/666110/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/10/666110/1"}}},"commit":{"parents":[{"commit":"2250c537e8cdee123138bb51c75d9daea2617d65","subject":"Merge \"Updated from OpenStack Ansible Tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2250c537e8cdee123138bb51c75d9daea2617d65"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:15:57.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-06-18 18:15:57.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: Ia6feeb3a386026e88b9557ecc9448f11084f2375\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/03b0aaf019d44e2fff3658a65c4f5cf503f61d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/03b0aaf019d44e2fff3658a65c4f5cf503f61d35"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
