)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~668057","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Id76a64dccc3a48bbc1e7b95337cd99acdc0e6cf3","project":"openstack/openstack-ansible-os_keystone","branch":"master","hashtags":[],"change_id":"Id76a64dccc3a48bbc1e7b95337cd99acdc0e6cf3","subject":"Update distro package names for openSUSE Leap 15.1","status":"MERGED","created":"2019-06-27 21:28:39.000000000","updated":"2019-07-10 20:03:48.000000000","submitted":"2019-07-10 20:03:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"668057-1562789028382-31ef7618","meta_rev_id":"485ac9e975ad4d5464e59567b1fd5378ffab59d0","_number":668057,"virtual_id_number":668057,"owner":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-10 20:03:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-10 18:34:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-01 20:24:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-10 18:34:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-01 20:24:39.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-10 18:34:37.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-10 20:03:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"17c2e53f959942f5ec3e9dcf7e218a768712a2bb","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-27 21:28:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e507310fdb32040e022c896c6674b2858a132a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-27 23:21:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/57/668057/1/check/openstack-ansible-linters/1934bc8/ : SUCCESS in 6m 46s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/57/668057/1/check/openstack-ansible-deploy-aio_metal-debian-stable/580f008/ : SUCCESS in 57m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/57/668057/1/check/openstack-ansible-deploy-aio_metal-centos-7/9e44e96/ : SUCCESS in 1h 16m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/57/668057/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/9b47577/ : SUCCESS in 1h 09m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/57/668057/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/d3d291c/ : SUCCESS in 1h 01m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/57/668057/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/87361f1/ : SUCCESS in 1h 14m 16s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/57/668057/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/584903e/ : SUCCESS in 51m 55s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/57/668057/1/check/openstack-tox-docs/25b5404/html/ : SUCCESS in 2m 43s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/57/668057/1/check/openstack-ansible-uw_apache-centos-7/01ba4cc/ : SUCCESS in 39m 01s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/57/668057/1/check/openstack-ansible-uw_apache-ubuntu-bionic/2519c07/ : SUCCESS in 34m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c8255330ebf5473a35da4e9227d274e403289d1e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 07:58:39.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"1ca72dce5c4b707ae197051538a1d8558685b0a0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-06-28 07:59:45.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bc530b6d0c8c0722a2d3b6b60ea4dc808e287d7b","author":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"date":"2019-06-28 08:42:37.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"9d64e150b0b3e7367385823978802ea195642cb2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 10:07:26.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/57/668057/3/check/openstack-ansible-linters/9c7e2f8/ : SUCCESS in 4m 39s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/57/668057/3/check/openstack-ansible-deploy-aio_metal-debian-stable/6f09278/ : SUCCESS in 1h 02m 33s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/57/668057/3/check/openstack-ansible-deploy-aio_metal-centos-7/64ab909/ : SUCCESS in 1h 18m 45s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/57/668057/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/f1a7a56/ : SUCCESS in 1h 03m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/57/668057/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/5fa3f8c/ : SUCCESS in 1h 02m 13s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/57/668057/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/259eeda/ : SUCCESS in 59m 49s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/57/668057/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/a0d9aad/ : SUCCESS in 55m 12s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/57/668057/3/check/openstack-tox-docs/9501254/html/ : SUCCESS in 2m 47s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/57/668057/3/check/openstack-ansible-uw_apache-centos-7/b48327c/ : SUCCESS in 39m 17s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/57/668057/3/check/openstack-ansible-uw_apache-ubuntu-bionic/3c2178d/ : SUCCESS in 34m 56s","accounts_in_message":[],"_revision_number":3},{"id":"801bcc403f98bd4e74afad871baf2c4ace5f5f71","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-01 20:24:39.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"14a9c18fe57532af11e171c186e894e82e83ad08","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-10 18:34:37.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"71fa4058298a5338e9642722c03e416e72bb5ed2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 18:36:29.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"698a72505665edaf638a451bc597b546a81ad6dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 20:03:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"485ac9e975ad4d5464e59567b1fd5378ffab59d0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 20:03:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/57/668057/3/gate/openstack-ansible-linters/ce99e05/ : SUCCESS in 6m 45s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/57/668057/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/a5eac9c/ : SUCCESS in 1h 04m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/57/668057/3/gate/openstack-ansible-deploy-aio_metal-centos-7/022690f/ : SUCCESS in 1h 23m 20s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/57/668057/3/gate/openstack-ansible-deploy-aio_metal-opensuse-15/4b1b5b4/ : SUCCESS in 1h 14m 05s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/57/668057/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/46ce488/ : SUCCESS in 1h 00m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/57/668057/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/024ccbb/ : SUCCESS in 1h 00m 56s\n- openstack-tox-docs http://logs.openstack.org/57/668057/3/gate/openstack-tox-docs/63d87f9/html/ : SUCCESS in 4m 06s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/57/668057/3/gate/openstack-ansible-uw_apache-centos-7/9ee9dc5/ : SUCCESS in 40m 13s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/57/668057/3/gate/openstack-ansible-uw_apache-ubuntu-bionic/63c6975/ : SUCCESS in 30m 50s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e968c4487e7b85d9d765885d8cd72e4969ea37e6","revisions":{"122caad778d18e966824a299abdb37e4a95c8751":{"kind":"REWORK","_number":1,"created":"2019-06-27 21:28:39.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/57/668057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/57/668057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/57/668057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/57/668057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/57/668057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/57/668057/1"}}},"commit":{"parents":[{"commit":"17b276a6eb281c43929a29a30592bd1ef19a7309","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/17b276a6eb281c43929a29a30592bd1ef19a7309"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-27 21:28:18.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-27 21:28:30.000000000","tz":120},"subject":"Switch to openSUSE Leap 15.1","message":"Switch to openSUSE Leap 15.1\n\nDepends-On: https://review.opendev.org/666058\nChange-Id: Id76a64dccc3a48bbc1e7b95337cd99acdc0e6cf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/122caad778d18e966824a299abdb37e4a95c8751"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/122caad778d18e966824a299abdb37e4a95c8751"}]},"branch":"refs/heads/master"},"362f45005b4489e70ea7a27ef10a4fc2ed4f90d2":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-06-28 07:58:39.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/57/668057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/57/668057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/57/668057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/57/668057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/57/668057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/57/668057/2"}}},"commit":{"parents":[{"commit":"17b276a6eb281c43929a29a30592bd1ef19a7309","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/17b276a6eb281c43929a29a30592bd1ef19a7309"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-27 21:28:18.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-06-28 07:58:33.000000000","tz":0},"subject":"Update distro package names for openSUSE Leap 15.1","message":"Update distro package names for openSUSE Leap 15.1\n\nRemove unecessary packages and update names where required.\n\nDepends-On: https://review.opendev.org/666058\nChange-Id: Id76a64dccc3a48bbc1e7b95337cd99acdc0e6cf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/362f45005b4489e70ea7a27ef10a4fc2ed4f90d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/362f45005b4489e70ea7a27ef10a4fc2ed4f90d2"}]},"branch":"refs/heads/master"},"e968c4487e7b85d9d765885d8cd72e4969ea37e6":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2019-06-28 08:42:37.000000000","uploader":{"_account_id":6593,"name":"Dirk Mueller","email":"dirk@dmllr.de","username":"dmllr"},"ref":"refs/changes/57/668057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/57/668057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/57/668057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/57/668057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/57/668057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/57/668057/3"}}},"commit":{"parents":[{"commit":"17b276a6eb281c43929a29a30592bd1ef19a7309","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/17b276a6eb281c43929a29a30592bd1ef19a7309"}]}],"author":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-27 21:28:18.000000000","tz":120},"committer":{"name":"Dirk Mueller","email":"dirk@dmllr.de","date":"2019-06-28 08:42:29.000000000","tz":120},"subject":"Update distro package names for openSUSE Leap 15.1","message":"Update distro package names for openSUSE Leap 15.1\n\nRemove unecessary packages and update names where required.\n\nChange-Id: Id76a64dccc3a48bbc1e7b95337cd99acdc0e6cf3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e968c4487e7b85d9d765885d8cd72e4969ea37e6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e968c4487e7b85d9d765885d8cd72e4969ea37e6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
