)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~670292","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I4824faedd1c663ac004a9e2674988c565f4cc27f","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"bind-to-mgmt","hashtags":[],"change_id":"I4824faedd1c663ac004a9e2674988c565f4cc27f","subject":"Add default bind addresses for nginx, apache and uwsgi","status":"MERGED","created":"2019-07-11 12:29:19.000000000","updated":"2019-07-22 14:40:06.000000000","submitted":"2019-07-22 14:40:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"670292-1563806406740-f7d455a8","meta_rev_id":"77f21b194d3e13c4a2471ce8ccf5c4ebd5682513","_number":670292,"virtual_id_number":670292,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2019-07-22 14:40:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2019-07-22 12:57:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-22 12:57:14.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-22 13:12:19.000000000","_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-11 18:08:49.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2019-07-22 13:12:19.000000000","updated_by":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"reviewer":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"state":"REVIEWER"},{"updated":"2019-07-22 14:40:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"705740d04557437c2450cbb76ec03c9fc3978d66","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-11 12:29:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"df8cdee7e2b0f602715b378c85e5559f15aa5796","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 13:06:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/92/670292/1/check/openstack-ansible-linters/c0d3c1e/ : SUCCESS in 5m 18s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/92/670292/1/check/openstack-ansible-deploy-aio_metal-debian-stable/68996cc/ : FAILURE in 25m 23s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/92/670292/1/check/openstack-ansible-deploy-aio_metal-centos-7/a07e0bb/ : FAILURE in 31m 00s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/92/670292/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/49deec7/ : FAILURE in 26m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/92/670292/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/8e9ff5f/ : FAILURE in 26m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/92/670292/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/f025a05/ : FAILURE in 25m 04s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/92/670292/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/5b69f19/ : FAILURE in 24m 06s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/92/670292/1/check/openstack-tox-docs/b9d8ab3/html/ : SUCCESS in 2m 34s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/92/670292/1/check/openstack-ansible-uw_apache-centos-7/8455ccc/ : FAILURE in 27m 15s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/92/670292/1/check/openstack-ansible-uw_apache-ubuntu-bionic/798eabf/ : FAILURE in 20m 04s","accounts_in_message":[],"_revision_number":1},{"id":"11184ce5e82f0f476e5e16631bfea957ea84ca9a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-11 13:07:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"edc315cc5ad4c47b9f36549a77be8b8841ddfe2e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-11 13:10:51.000000000","message":"Topic set to bind-to-mgmt","accounts_in_message":[],"_revision_number":2},{"id":"a6331977efdeb7923d9338ed40787a53b9f95244","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-11 15:04:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/92/670292/2/check/openstack-ansible-linters/96bde99/ : SUCCESS in 8m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/92/670292/2/check/openstack-ansible-deploy-aio_metal-debian-stable/5cbd648/ : SUCCESS in 1h 18m 52s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/92/670292/2/check/openstack-ansible-deploy-aio_metal-centos-7/02fcd09/ : SUCCESS in 1h 13m 06s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/92/670292/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/6cbfc33/ : SUCCESS in 1h 12m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/92/670292/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/2776b51/ : SUCCESS in 57m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/92/670292/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/1a2be45/ : SUCCESS in 59m 41s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/92/670292/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/bf631b7/ : FAILURE in 29m 21s (non-voting)\n- openstack-tox-docs http://logs.openstack.org/92/670292/2/check/openstack-tox-docs/24010af/html/ : SUCCESS in 3m 03s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/92/670292/2/check/openstack-ansible-uw_apache-centos-7/9a40868/ : SUCCESS in 34m 36s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/92/670292/2/check/openstack-ansible-uw_apache-ubuntu-bionic/dd78460/ : SUCCESS in 31m 03s","accounts_in_message":[],"_revision_number":2},{"id":"a99a39346218c4e587b8c9e157a8e514c2243a7c","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-07-11 18:08:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"16c82a207e60d162133557b8ff790769c58bc087","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-12 16:01:41.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"93576afc0bd3556a2b70131a78c1e801a0fa181a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-15 12:37:37.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\ndoes anything here actually bind to ansible_default_ipv4.address? I see the test in the handler using that as the wait_for host but no further than that.","accounts_in_message":[],"_revision_number":2},{"id":"2bdd690a2227c22d54cb13d5f6cd6df767908bbe","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-22 12:46:20.000000000","message":"Removed Code-Review-1 by Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"30322d2653d2acef3fa31f3bb4073b648965015a","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-22 12:46:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLet\u0027s make that happen","accounts_in_message":[],"_revision_number":2},{"id":"d463b0d56eadfe6556e6332dd846da52fb195569","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-22 12:46:34.000000000","message":"Removed Workflow+1 by Jean-Philippe Evrard \u003cjean-philippe@evrard.me\u003e\n","accounts_in_message":[],"_revision_number":2},{"id":"5ff6337aa8f53641fb9229994ddb254c140175d7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-22 12:57:14.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"dd6af55e9ca80f343ea82030d3c62bae1445bf85","author":{"_account_id":17068,"name":"Jean-Philippe Evrard","email":"openstack@a.spamming.party","username":"evrardjp"},"date":"2019-07-22 13:12:19.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"24a4bc6adad5ef4ad207c374ab1c1dbc7940cdec","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 13:12:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"23a77eaa186ed02448b02e542d49168807c28936","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 14:40:06.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/92/670292/3/gate/openstack-ansible-linters/e866627/ : SUCCESS in 6m 27s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/92/670292/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/53c3962/ : SUCCESS in 1h 18m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/92/670292/3/gate/openstack-ansible-deploy-aio_metal-centos-7/d2cc695/ : SUCCESS in 1h 21m 12s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/92/670292/3/gate/openstack-ansible-deploy-aio_metal-opensuse-15/f0fa29b/ : SUCCESS in 1h 12m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/92/670292/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/3099743/ : SUCCESS in 1h 04m 15s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/92/670292/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/7f5e9fb/ : SUCCESS in 1h 02m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/92/670292/3/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/3ae1b71/ : SUCCESS in 46m 01s\n- openstack-tox-docs http://logs.openstack.org/92/670292/3/gate/openstack-tox-docs/2b6abd3/html/ : SUCCESS in 3m 02s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/92/670292/3/gate/openstack-ansible-uw_apache-centos-7/0774a06/ : SUCCESS in 39m 39s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/92/670292/3/gate/openstack-ansible-uw_apache-ubuntu-bionic/6c1ae02/ : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":3},{"id":"77f21b194d3e13c4a2471ce8ccf5c4ebd5682513","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 14:40:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e3a9237b83c99547c5eaf976fc59fda9ad302f6e","revisions":{"5b23db12f62e8bdeb91fc2cb6593007357ce512e":{"kind":"REWORK","_number":1,"created":"2019-07-11 12:29:19.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/92/670292/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/92/670292/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/670292/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/670292/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/670292/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/670292/1"}}},"commit":{"parents":[{"commit":"5132d6a83999fe5f76f62a06ce0d296f427125ab","subject":"Merge \"Update distro package names for openSUSE Leap 15.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5132d6a83999fe5f76f62a06ce0d296f427125ab"}]}],"author":{"name":"Your Name","email":"you@example.com","date":"2019-07-11 12:23:55.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 12:29:00.000000000","tz":60},"subject":"Add default bind addresses for nginx, apache and uwsgi","message":"Add default bind addresses for nginx, apache and uwsgi\n\nThese can be overriden to bind to the actual management network IP\nin a real deployment\n\nChange-Id: I4824faedd1c663ac004a9e2674988c565f4cc27f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5b23db12f62e8bdeb91fc2cb6593007357ce512e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5b23db12f62e8bdeb91fc2cb6593007357ce512e"}]},"branch":"refs/heads/master"},"deaeab716e26262f9fab5689b91d9d6893bf5fd4":{"kind":"REWORK","_number":2,"created":"2019-07-11 13:07:43.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/92/670292/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/92/670292/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/670292/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/670292/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/670292/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/670292/2"}}},"commit":{"parents":[{"commit":"5132d6a83999fe5f76f62a06ce0d296f427125ab","subject":"Merge \"Update distro package names for openSUSE Leap 15.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5132d6a83999fe5f76f62a06ce0d296f427125ab"}]}],"author":{"name":"Your Name","email":"you@example.com","date":"2019-07-11 12:23:55.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-11 13:07:39.000000000","tz":0},"subject":"Add default bind addresses for nginx, apache and uwsgi","message":"Add default bind addresses for nginx, apache and uwsgi\n\nThese can be overriden to bind to the actual management network IP\nin a real deployment\n\nChange-Id: I4824faedd1c663ac004a9e2674988c565f4cc27f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/deaeab716e26262f9fab5689b91d9d6893bf5fd4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/deaeab716e26262f9fab5689b91d9d6893bf5fd4"}]},"branch":"refs/heads/master"},"e3a9237b83c99547c5eaf976fc59fda9ad302f6e":{"kind":"NO_CHANGE","_number":3,"created":"2019-07-22 12:57:14.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/92/670292/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/92/670292/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/670292/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/670292/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/670292/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/92/670292/3"}}},"commit":{"parents":[{"commit":"5132d6a83999fe5f76f62a06ce0d296f427125ab","subject":"Merge \"Update distro package names for openSUSE Leap 15.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5132d6a83999fe5f76f62a06ce0d296f427125ab"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-22 12:56:54.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-07-22 12:56:54.000000000","tz":60},"subject":"Add default bind addresses for nginx, apache and uwsgi","message":"Add default bind addresses for nginx, apache and uwsgi\n\nThese can be overriden to bind to the actual management network IP\nin a real deployment\n\nChange-Id: I4824faedd1c663ac004a9e2674988c565f4cc27f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e3a9237b83c99547c5eaf976fc59fda9ad302f6e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e3a9237b83c99547c5eaf976fc59fda9ad302f6e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
