)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~671603","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I2fabfc3c182a51ba837010651e359e4f2017f844","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I2fabfc3c182a51ba837010651e359e4f2017f844","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-18 22:18:45.000000000","updated":"2019-07-20 17:40:02.000000000","submitted":"2019-07-20 17:40:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671603-1563644402190-a91269ad","meta_rev_id":"ff5f4d136fb7d2fcee269c38907e19f3689e4e19","_number":671603,"virtual_id_number":671603,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":2,"date":"2019-07-20 17:40:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-20 16:14:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-20 16:14:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-20 16:14:42.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-20 17:40:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"00e0fe5d312a9f6f406e209ad47661a0d5e13eaf","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-18 22:18:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c140efd3d096cd73ff3c35fdbbeb27bf11e4a9e7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-18 23:48:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/03/671603/1/check/openstack-ansible-linters/dc55308/ : SUCCESS in 5m 22s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/03/671603/1/check/openstack-ansible-deploy-aio_metal-debian-stable/3857f11/ : SUCCESS in 1h 14m 28s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/03/671603/1/check/openstack-ansible-deploy-aio_metal-centos-7/df72c03/ : SUCCESS in 1h 27m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/03/671603/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/d2d4424/ : FAILURE in 1h 11m 04s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/03/671603/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/149cc38/ : SUCCESS in 1h 09m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/03/671603/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/41fe88f/ : SUCCESS in 55m 39s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/03/671603/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/14eff03/ : FAILURE in 47m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/03/671603/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/1168ac2/ : SUCCESS in 47m 52s\n- openstack-tox-docs http://logs.openstack.org/03/671603/1/check/openstack-tox-docs/a5350e7/html/ : SUCCESS in 4m 05s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/03/671603/1/check/openstack-ansible-uw_apache-centos-7/87a2937/ : SUCCESS in 38m 36s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/03/671603/1/check/openstack-ansible-uw_apache-ubuntu-bionic/7d461b0/ : SUCCESS in 36m 17s","accounts_in_message":[],"_revision_number":1},{"id":"3d4d7f3ad5f3a07ddad6ac2cc057a49bad6c346f","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-19 18:22:54.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"16d04a554471a2fa2e94be58875c5418316253f9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 21:20:44.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- requirements-check http://logs.openstack.org/03/671603/2/check/requirements-check/31e9dfe/ : SUCCESS in 2m 42s\n- openstack-ansible-linters http://logs.openstack.org/03/671603/2/check/openstack-ansible-linters/d52f6a2/ : SUCCESS in 9m 48s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/03/671603/2/check/openstack-ansible-deploy-aio_metal-debian-stable/47097c0/ : SUCCESS in 1h 08m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/03/671603/2/check/openstack-ansible-deploy-aio_metal-centos-7/fdf1fbf/ : POST_FAILURE in 1h 48m 25s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/03/671603/2/check/openstack-ansible-deploy-aio_metal-opensuse-15/7755faf/ : SUCCESS in 1h 13m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/03/671603/2/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b8e5f3b/ : SUCCESS in 58m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/03/671603/2/check/openstack-ansible-deploy-aio_distro_metal-centos-7/5e4093d/ : SUCCESS in 1h 09m 10s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/03/671603/2/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/ec4fbb3/ : FAILURE in 49m 40s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/03/671603/2/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/ccd8c12/ : SUCCESS in 49m 28s\n- openstack-tox-docs http://logs.openstack.org/03/671603/2/check/openstack-tox-docs/7b9fc16/ : POST_FAILURE in 3m 26s\n- build-openstack-releasenotes http://logs.openstack.org/03/671603/2/check/build-openstack-releasenotes/0e27480/html/ : SUCCESS in 4m 26s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/03/671603/2/check/openstack-ansible-uw_apache-centos-7/d0acb2b/ : SUCCESS in 34m 47s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/03/671603/2/check/openstack-ansible-uw_apache-ubuntu-bionic/6121ea4/ : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":2},{"id":"8d5bc8270d9d97cd490be0c80188bbd4c0e58d31","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-20 09:33:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e1eaf46310a320a3274cfbabcc7f8238dedfe268","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 10:54:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/03/671603/3/check/requirements-check/c27e903/ : SUCCESS in 3m 15s\n- openstack-ansible-linters http://logs.openstack.org/03/671603/3/check/openstack-ansible-linters/515b587/ : SUCCESS in 7m 14s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/03/671603/3/check/openstack-ansible-deploy-aio_metal-debian-stable/0a5acc9/ : SUCCESS in 56m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/03/671603/3/check/openstack-ansible-deploy-aio_metal-centos-7/f44e150/ : SUCCESS in 1h 14m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/03/671603/3/check/openstack-ansible-deploy-aio_metal-opensuse-15/c19a9b4/ : SUCCESS in 1h 05m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/03/671603/3/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/441e8fc/ : SUCCESS in 59m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/03/671603/3/check/openstack-ansible-deploy-aio_distro_metal-centos-7/afb86fd/ : SUCCESS in 53m 47s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/03/671603/3/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/1d18b37/ : FAILURE in 44m 57s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/03/671603/3/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/b1fa06f/ : SUCCESS in 41m 27s\n- openstack-tox-docs http://logs.openstack.org/03/671603/3/check/openstack-tox-docs/7453c12/html/ : SUCCESS in 2m 18s\n- build-openstack-releasenotes http://logs.openstack.org/03/671603/3/check/build-openstack-releasenotes/7bf3651/html/ : SUCCESS in 3m 44s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/03/671603/3/check/openstack-ansible-uw_apache-centos-7/d260881/ : SUCCESS in 35m 30s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/03/671603/3/check/openstack-ansible-uw_apache-ubuntu-bionic/136e399/ : SUCCESS in 34m 43s","accounts_in_message":[],"_revision_number":3},{"id":"f635292029206c26d826a85e7c8f080ad589db20","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-20 16:14:42.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a70aff23d0bd3a0fdebd324507459ddad824bf4b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 16:14:53.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"00c0f814234a6c63699f8996a9aaba1676f6da1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 17:40:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/03/671603/3/gate/requirements-check/8ddd27f/ : SUCCESS in 3m 35s\n- openstack-ansible-linters http://logs.openstack.org/03/671603/3/gate/openstack-ansible-linters/99b53b0/ : SUCCESS in 9m 36s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/03/671603/3/gate/openstack-ansible-deploy-aio_metal-debian-stable/4342f46/ : SUCCESS in 1h 05m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/03/671603/3/gate/openstack-ansible-deploy-aio_metal-centos-7/eefd208/ : SUCCESS in 1h 20m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/03/671603/3/gate/openstack-ansible-deploy-aio_metal-opensuse-15/d589a19/ : SUCCESS in 1h 16m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/03/671603/3/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9796f9c/ : SUCCESS in 1h 10m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/03/671603/3/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/79daedd/ : SUCCESS in 1h 01m 06s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/03/671603/3/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/79563d4/ : SUCCESS in 1h 00m 19s\n- openstack-tox-docs http://logs.openstack.org/03/671603/3/gate/openstack-tox-docs/3ae66c2/html/ : SUCCESS in 4m 22s\n- build-openstack-releasenotes http://logs.openstack.org/03/671603/3/gate/build-openstack-releasenotes/7eda768/html/ : SUCCESS in 3m 21s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/03/671603/3/gate/openstack-ansible-uw_apache-centos-7/ae1cf44/ : SUCCESS in 39m 37s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/03/671603/3/gate/openstack-ansible-uw_apache-ubuntu-bionic/5e42317/ : SUCCESS in 34m 17s","accounts_in_message":[],"_revision_number":3},{"id":"455511a5c58777255bfc1f90c2b198629abf978d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-20 17:40:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"0c516c26dae25a6d0e3652849f3aa63d1138f855","revisions":{"a19d0fb9129eb2bd38508429b28fed334922152c":{"kind":"REWORK","_number":1,"created":"2019-07-18 22:18:45.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/671603/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/03/671603/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/671603/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/671603/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/671603/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/671603/1"}}},"commit":{"parents":[{"commit":"5132d6a83999fe5f76f62a06ce0d296f427125ab","subject":"Merge \"Update distro package names for openSUSE Leap 15.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5132d6a83999fe5f76f62a06ce0d296f427125ab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:18:44.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-18 22:18:44.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2fabfc3c182a51ba837010651e359e4f2017f844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a19d0fb9129eb2bd38508429b28fed334922152c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a19d0fb9129eb2bd38508429b28fed334922152c"}]},"branch":"refs/heads/master"},"c1bcb3fed441b757fb5aee771605eeae4eca6270":{"kind":"REWORK","_number":2,"created":"2019-07-19 18:22:54.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/671603/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/03/671603/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/671603/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/671603/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/671603/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/671603/2"}}},"commit":{"parents":[{"commit":"5132d6a83999fe5f76f62a06ce0d296f427125ab","subject":"Merge \"Update distro package names for openSUSE Leap 15.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5132d6a83999fe5f76f62a06ce0d296f427125ab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-19 18:22:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-19 18:22:50.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2fabfc3c182a51ba837010651e359e4f2017f844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c1bcb3fed441b757fb5aee771605eeae4eca6270"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c1bcb3fed441b757fb5aee771605eeae4eca6270"}]},"branch":"refs/heads/master"},"0c516c26dae25a6d0e3652849f3aa63d1138f855":{"kind":"REWORK","_number":3,"created":"2019-07-20 09:33:39.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/03/671603/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/03/671603/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/671603/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/671603/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/671603/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/671603/3"}}},"commit":{"parents":[{"commit":"5132d6a83999fe5f76f62a06ce0d296f427125ab","subject":"Merge \"Update distro package names for openSUSE Leap 15.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/5132d6a83999fe5f76f62a06ce0d296f427125ab"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-20 09:33:35.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-20 09:33:35.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I2fabfc3c182a51ba837010651e359e4f2017f844\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0c516c26dae25a6d0e3652849f3aa63d1138f855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0c516c26dae25a6d0e3652849f3aa63d1138f855"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
