)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~671719","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fstein~Ic92c2c8172e4f2f9eca2b97e1c1d8cbf09dd167d","project":"openstack/openstack-ansible-os_keystone","branch":"stable/stein","topic":"static_imports-stable/stein","hashtags":[],"change_id":"Ic92c2c8172e4f2f9eca2b97e1c1d8cbf09dd167d","subject":"Convert dynamic includes to static imports","status":"MERGED","created":"2019-07-19 13:10:31.000000000","updated":"2019-07-19 18:11:23.000000000","submitted":"2019-07-19 18:11:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"671719-1563559883398-ef2d6f53","meta_rev_id":"fba83973ceef50eeed98fe1daec936ed41ae1486","_number":671719,"virtual_id_number":671719,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-19 18:11:23.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-19 16:08:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-19 16:30:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-19 16:30:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-19 16:08:30.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-19 16:30:41.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-19 18:11:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"45fb53d8f0aa91763d73e640e0e656a27b47cc30","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-07-19 13:10:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1eac81f08b14a7cfb62563aab139fb4ff7a34758","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 14:31:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/19/671719/1/check/openstack-ansible-linters/6b37aa6/ : SUCCESS in 7m 00s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/19/671719/1/check/openstack-ansible-deploy-aio_metal-debian-stable/1e2c946/ : SUCCESS in 1h 05m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/19/671719/1/check/openstack-ansible-deploy-aio_metal-centos-7/7f2e164/ : SUCCESS in 1h 12m 09s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/19/671719/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/958009a/ : SUCCESS in 1h 02m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/19/671719/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/9a7f7a9/ : SUCCESS in 56m 24s\n- openstack-tox-docs http://logs.openstack.org/19/671719/1/check/openstack-tox-docs/fb67f53/html/ : SUCCESS in 3m 18s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/19/671719/1/check/openstack-ansible-uw_apache-centos-7/b62abaf/ : SUCCESS in 38m 08s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/19/671719/1/check/openstack-ansible-uw_apache-ubuntu-bionic/c995780/ : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":1},{"id":"fff7d9e69686c80cf51ec1505ca86dd5fe59b34b","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-19 16:08:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2914bd6cee65f6a17cc91282bec538cfb10673b4","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-19 16:30:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f311f13486c86537a0c3a021cc77997a2dbf9e19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 16:30:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"016f42eaf41cf486adf358a26064624cdeac5031","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:11:23.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"bf87ceab3a545cb9c2074e13ff8886c7b95aa0e2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-19 18:11:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/19/671719/1/gate/openstack-ansible-linters/7a18c4c/ : SUCCESS in 7m 42s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/19/671719/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/bfb98c7/ : SUCCESS in 54m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/19/671719/1/gate/openstack-ansible-deploy-aio_metal-centos-7/4fba52b/ : SUCCESS in 1h 25m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/19/671719/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/1e4407c/ : SUCCESS in 52m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/19/671719/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/839e3f1/ : SUCCESS in 1h 02m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/19/671719/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/a35cfac/ : SUCCESS in 39m 47s\n- openstack-tox-docs http://logs.openstack.org/19/671719/1/gate/openstack-tox-docs/a9641f7/html/ : SUCCESS in 2m 56s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/19/671719/1/gate/openstack-ansible-uw_apache-centos-7/6920e43/ : SUCCESS in 38m 02s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/19/671719/1/gate/openstack-ansible-uw_apache-ubuntu-bionic/8c76536/ : SUCCESS in 33m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f14c2ac96f884160c4353d014f748576697825a0","revisions":{"f14c2ac96f884160c4353d014f748576697825a0":{"kind":"REWORK","_number":1,"created":"2019-07-19 13:10:31.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/19/671719/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/19/671719/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/19/671719/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/19/671719/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/19/671719/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/19/671719/1"}}},"commit":{"parents":[{"commit":"9278652165e432aa0c8cbc9829223ea3ffbceb8f","subject":"Merge \"Execute the keystone db setup earlier in the process\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9278652165e432aa0c8cbc9829223ea3ffbceb8f"}]}],"author":{"name":"Dmitriy Rabotjagov","email":"dmitriy.r@sitevalley.com","date":"2019-06-13 13:46:46.000000000","tz":180},"committer":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2019-07-19 13:10:15.000000000","tz":180},"subject":"Convert dynamic includes to static imports","message":"Convert dynamic includes to static imports\n\nWhen task/role files are included using include_tasks, tags are not\npassed to the included tasks. As a result, tags like neutron-config\ndo not have the intended effect. This patch changes include_tasks\nto import_tasks for all cases where dynamic vars or loops are not used\nso that tags are properly handled.\n\nReference -\nhttps://docs.ansible.com/ansible/latest/user_guide/playbooks_reuse.html\nhttps://bugs.launchpad.net/openstack-ansible/+bug/1815043\n\nChange-Id: Ic92c2c8172e4f2f9eca2b97e1c1d8cbf09dd167d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f14c2ac96f884160c4353d014f748576697825a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f14c2ac96f884160c4353d014f748576697825a0"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
