)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~672040","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ic2e3bc66161217b2884001018a706dbe2187c498","project":"openstack/openstack-ansible-os_keystone","branch":"master","hashtags":[],"change_id":"Ic2e3bc66161217b2884001018a706dbe2187c498","subject":"Remove nginx default conf","status":"MERGED","created":"2019-07-22 13:20:31.000000000","updated":"2019-07-23 09:00:29.000000000","submitted":"2019-07-23 09:00:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"672040-1563872429340-a91269ad","meta_rev_id":"39772973925022581c519e6881136f225a0235b3","_number":672040,"virtual_id_number":672040,"owner":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-23 09:00:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":2,"date":"2019-07-22 13:59:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-07-22 15:44:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"date":"2019-07-22 13:22:34.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-07-22 15:44:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-22 13:22:34.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-07-22 13:59:57.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-22 15:44:43.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-07-23 09:00:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d25d9fb142c52523100b2c1e3a3251d9913b1c1d","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2019-07-22 13:20:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4020549c3b666d9c5c10cc9d71aecf436afd0f38","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-22 13:59:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d27d0ecca37dab351a5af3aeae34ae0295cba9ef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 14:56:44.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/40/672040/1/check/openstack-ansible-linters/c345f5a/ : SUCCESS in 7m 02s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_metal-debian-stable/3d30439/ : SUCCESS in 1h 04m 50s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_metal-centos-7/e76cafd/ : SUCCESS in 1h 29m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/81d1b3f/ : SUCCESS in 1h 03m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/ff4bc61/ : SUCCESS in 1h 08m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/ea569b3/ : SUCCESS in 1h 06m 20s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/a596fb1/ : FAILURE in 48m 08s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/cbef46f/ : SUCCESS in 43m 17s\n- openstack-tox-docs http://logs.openstack.org/40/672040/1/check/openstack-tox-docs/b1b5b1b/html/ : SUCCESS in 3m 23s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/40/672040/1/check/openstack-ansible-uw_apache-centos-7/aa233fa/ : SUCCESS in 35m 52s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/40/672040/1/check/openstack-ansible-uw_apache-ubuntu-bionic/3718c63/ : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":1},{"id":"78b7ef4a31c3f6bcf10394ed0f24397a4394ffcb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-22 15:44:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"324d0311e68118c4141c5ad5e89470487b3611a1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 15:44:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"43eca3794958f1564f6d90c7f1e9eff62750f9d3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 17:24:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/40/672040/1/gate/openstack-ansible-linters/fdc9639/ : SUCCESS in 6m 49s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/aa22b7e/ : SUCCESS in 1h 06m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_metal-centos-7/67248da/ : SUCCESS in 1h 27m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/380682f/ : SUCCESS in 1h 26m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/f92e2cd/ : SUCCESS in 1h 09m 43s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/c752396/ : SUCCESS in 1h 16m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/6f45604/ : SUCCESS in 50m 00s\n- openstack-tox-docs http://logs.openstack.org/40/672040/1/gate/openstack-tox-docs/0d1532d/html/ : SUCCESS in 3m 21s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/40/672040/1/gate/openstack-ansible-uw_apache-centos-7/8e9619f/ : FAILURE in 9m 59s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/40/672040/1/gate/openstack-ansible-uw_apache-ubuntu-bionic/0829f20/ : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"efbe448f10adb3d89df77fafc953b2b7e2db93a3","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-22 20:43:59.000000000","message":"Patch Set 1:\n\nrecheck\n\nException: [AbstractCommand.cc:351] errorCode\u003d22 URI\u003dhttps://github.com/CentOS/sig-cloud-instance-images/raw/CentOS-7/docker/centos-7-docker.tar.xz\n  -\u003e [HttpSkipResponseCommand.cc:240] errorCode\u003d22 The response status is not successful. status\u003d500","accounts_in_message":[],"_revision_number":1},{"id":"3e922bc512b276861666edd44d0202402c1af946","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 22:14:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/40/672040/1/check/openstack-ansible-linters/9edc0af/ : SUCCESS in 9m 57s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_metal-debian-stable/95ad1d1/ : SUCCESS in 1h 09m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_metal-centos-7/50c8cf8/ : SUCCESS in 1h 16m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/6e12200/ : SUCCESS in 1h 09m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/b9e7118/ : SUCCESS in 1h 03m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/e581e3e/ : SUCCESS in 1h 00m 03s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/92cc5cd/ : FAILURE in 45m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/bc9d6db/ : SUCCESS in 50m 46s\n- openstack-tox-docs http://logs.openstack.org/40/672040/1/check/openstack-tox-docs/25f4592/html/ : SUCCESS in 3m 33s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/40/672040/1/check/openstack-ansible-uw_apache-centos-7/700bbbf/ : SUCCESS in 41m 30s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/40/672040/1/check/openstack-ansible-uw_apache-ubuntu-bionic/9170dcd/ : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":1},{"id":"a3409976e63f0be0274cb46213d49cdc758be3bb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 22:14:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9f8567efe76ba5c464868f06ca7761e4b25a06fd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-22 23:50:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters http://logs.openstack.org/40/672040/1/gate/openstack-ansible-linters/774c046/ : SUCCESS in 10m 32s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/fdb8064/ : FAILURE in 53m 45s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_metal-centos-7/32cd45a/ : FAILURE in 16m 22s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/385bac3/ : SUCCESS in 1h 03m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/9a481da/ : SUCCESS in 1h 07m 20s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/15b977c/ : SUCCESS in 1h 05m 37s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/5455b82/ : SUCCESS in 49m 04s\n- openstack-tox-docs http://logs.openstack.org/40/672040/1/gate/openstack-tox-docs/6d0667b/html/ : SUCCESS in 3m 17s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/40/672040/1/gate/openstack-ansible-uw_apache-centos-7/b1783f2/ : SUCCESS in 39m 04s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/40/672040/1/gate/openstack-ansible-uw_apache-ubuntu-bionic/ff88e14/ : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2c689d5ab089b9f809018f21db2a3920cdfcecb0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-07-23 05:22:04.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"cabcaa71125db117f597452c07abbc578ec8fd35","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 07:18:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/40/672040/1/check/openstack-ansible-linters/74e8cba/ : SUCCESS in 5m 37s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_metal-debian-stable/9955108/ : SUCCESS in 1h 14m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_metal-centos-7/a564040/ : SUCCESS in 1h 42m 02s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_metal-opensuse-15/367bd78/ : SUCCESS in 1h 23m 10s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_metal-ubuntu-bionic/04d62bb/ : SUCCESS in 1h 01m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_distro_metal-centos-7/add4709/ : SUCCESS in 1h 08m 28s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_distro_metal-opensuse-15/36b29ff/ : FAILURE in 54m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/40/672040/1/check/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/fcc6434/ : SUCCESS in 56m 11s\n- openstack-tox-docs http://logs.openstack.org/40/672040/1/check/openstack-tox-docs/e6e1d86/html/ : SUCCESS in 3m 55s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/40/672040/1/check/openstack-ansible-uw_apache-centos-7/42a5830/ : SUCCESS in 41m 15s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/40/672040/1/check/openstack-ansible-uw_apache-ubuntu-bionic/6ca4827/ : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"b24e3b2476261cfe140c17fca30715e5e8c6e921","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 07:18:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89c07f708a643ddf3017c0c124419858681f761c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 09:00:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters http://logs.openstack.org/40/672040/1/gate/openstack-ansible-linters/50dd75d/ : SUCCESS in 7m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_metal-debian-stable/9e4cb6c/ : SUCCESS in 1h 07m 13s\n- openstack-ansible-deploy-aio_metal-centos-7 http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_metal-centos-7/c8233fb/ : SUCCESS in 1h 37m 21s\n- openstack-ansible-deploy-aio_metal-opensuse-15 http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_metal-opensuse-15/0ce2817/ : SUCCESS in 1h 04m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_metal-ubuntu-bionic/dc2f3bc/ : SUCCESS in 1h 08m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_distro_metal-centos-7/08f789e/ : SUCCESS in 1h 12m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic http://logs.openstack.org/40/672040/1/gate/openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic/3fdb7f6/ : SUCCESS in 47m 11s\n- openstack-tox-docs http://logs.openstack.org/40/672040/1/gate/openstack-tox-docs/e8c89e3/html/ : SUCCESS in 2m 47s\n- openstack-ansible-uw_apache-centos-7 http://logs.openstack.org/40/672040/1/gate/openstack-ansible-uw_apache-centos-7/c84efb8/ : SUCCESS in 35m 45s\n- openstack-ansible-uw_apache-ubuntu-bionic http://logs.openstack.org/40/672040/1/gate/openstack-ansible-uw_apache-ubuntu-bionic/02408ae/ : SUCCESS in 35m 57s","accounts_in_message":[],"_revision_number":1},{"id":"39772973925022581c519e6881136f225a0235b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-23 09:00:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"46f3456cf17c728fc9e7acee538fe82f9d14c4b3","revisions":{"46f3456cf17c728fc9e7acee538fe82f9d14c4b3":{"kind":"REWORK","_number":1,"created":"2019-07-22 13:20:31.000000000","uploader":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"ref":"refs/changes/40/672040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/40/672040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/40/672040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/40/672040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/40/672040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/40/672040/1"}}},"commit":{"parents":[{"commit":"0c516c26dae25a6d0e3652849f3aa63d1138f855","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/0c516c26dae25a6d0e3652849f3aa63d1138f855"}]}],"author":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-07-22 13:18:57.000000000","tz":0},"committer":{"name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","date":"2019-07-22 13:18:57.000000000","tz":0},"subject":"Remove nginx default conf","message":"Remove nginx default conf\n\nThis patch make sure Default.conf from nginx\nis remove as it causes socket conflict when\nwe deploy it on metal. Also, we remove the\ntask created by kevin for Q -\u003e S deployments\nas it is not required to S -\u003e T upgrades.\n\nChange-Id: Ic2e3bc66161217b2884001018a706dbe2187c498\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/46f3456cf17c728fc9e7acee538fe82f9d14c4b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/46f3456cf17c728fc9e7acee538fe82f9d14c4b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
