)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~673643","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I04f938ead0be6a3ea62a14dfd5d6c4244dda88ef","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I04f938ead0be6a3ea62a14dfd5d6c4244dda88ef","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2019-07-30 19:40:27.000000000","updated":"2019-07-30 20:16:51.000000000","submitted":"2019-07-30 20:16:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"673643-1564517811816-fe339b59","meta_rev_id":"187e212ed658322d3fd8c887a3407da70e7e7588","_number":673643,"virtual_id_number":673643,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":2,"date":"2019-07-30 20:16:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-07-30 20:02:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-07-30 20:02:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-30 20:02:55.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-07-30 20:16:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0097d516d96bf6b5c4e8da501f8876f184db501a","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2019-07-30 19:40:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8a54b9c4c8f754da187c1222c86e554c404f752e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:00:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check http://logs.openstack.org/43/673643/1/check/requirements-check/62c0361/ : SUCCESS in 2m 36s\n- openstack-ansible-linters http://logs.openstack.org/43/673643/1/check/openstack-ansible-linters/b8270e6/ : SUCCESS in 9m 25s\n- openstack-tox-docs http://logs.openstack.org/43/673643/1/check/openstack-tox-docs/9ceaf5f/html/ : SUCCESS in 4m 02s\n- build-openstack-releasenotes http://logs.openstack.org/43/673643/1/check/build-openstack-releasenotes/7a45635/html/ : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ca0977f4b818a294af54b6e0f90e3c15bf414a0d","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-07-30 20:02:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"98bb040d9fea82b74100909e43d9a8cb93fb1316","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:03:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1e01cac711fd75a6e1b7b997f1997b0786f617f6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:16:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"4b7fde753be4beba59fdebb8e0c1f1f524af0ba2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-30 20:16:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check http://logs.openstack.org/43/673643/1/gate/requirements-check/bff602f/ : SUCCESS in 3m 39s\n- openstack-ansible-linters http://logs.openstack.org/43/673643/1/gate/openstack-ansible-linters/eb98044/ : SUCCESS in 9m 36s\n- openstack-tox-docs http://logs.openstack.org/43/673643/1/gate/openstack-tox-docs/fdfd042/html/ : SUCCESS in 4m 04s\n- build-openstack-releasenotes http://logs.openstack.org/43/673643/1/gate/build-openstack-releasenotes/5d47fde/html/ : SUCCESS in 4m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3fc75ea13d3e5dff5d6a5f3a1fc4e7c1849800ed","revisions":{"3fc75ea13d3e5dff5d6a5f3a1fc4e7c1849800ed":{"kind":"REWORK","_number":1,"created":"2019-07-30 19:40:27.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/43/673643/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/43/673643/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/43/673643/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/43/673643/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/43/673643/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/43/673643/1"}}},"commit":{"parents":[{"commit":"67a8a385977d94f3cc9a27e69245771506df7419","subject":"service_setup: Update services tasks","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/67a8a385977d94f3cc9a27e69245771506df7419"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 19:40:23.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2019-07-30 19:40:23.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I04f938ead0be6a3ea62a14dfd5d6c4244dda88ef\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3fc75ea13d3e5dff5d6a5f3a1fc4e7c1849800ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3fc75ea13d3e5dff5d6a5f3a1fc4e7c1849800ed"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
