)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~682478","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Icddedc71e7ab464372e8e07fdce4c6ce79013d5b","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"octavia-ci-issue","hashtags":[],"change_id":"Icddedc71e7ab464372e8e07fdce4c6ce79013d5b","subject":"[DNM] test only","status":"ABANDONED","created":"2019-09-16 20:43:48.000000000","updated":"2019-09-18 18:54:19.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1996585122126aa669e2f123a213148fe0c3b1c4","_number":682478,"virtual_id_number":682478,"owner":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":-1,"date":"2019-09-16 22:26:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"all":[{"value":-1,"date":"2019-09-16 20:44:03.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-09-16 22:26:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"205636d5c9f4c32dbf04f2428822b00aa0ce0247","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-16 20:43:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43a1dc68d5c32c4a94c2091805624d50c1ec97b0","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-16 20:44:03.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"e84080c6ddbe115cdedcda7e92ce4dfd5829d397","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-16 22:26:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e426ebd5ead34b7e9c33fe78ea4d6bb1 : SUCCESS in 6m 17s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c5a3300be2464561af6a18b51da20e49 : SUCCESS in 1h 16m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bf65e9427941493dad1ccc68ae81b596 : SUCCESS in 1h 36m 32s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/5341f079aa97408ea8c4be0b1ff4c4b9 : SUCCESS in 1h 20m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e854da9129234e2e99af3e9bc49f79a3 : SUCCESS in 1h 15m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/35a3e1646ff8420fa11f96177f6d8c53 : FAILURE in 30m 30s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7598c7094a9e4a259a2629f3867a8808 : FAILURE in 23m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b9db3b6a15347b1adeae9ca7b9b84e2 : SUCCESS in 1h 10m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21463a1ac421454284da60a3c5bfc08f : SUCCESS in 4m 15s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/ce6ede29a18d4bd5931aa338eb7e702c : FAILURE in 27m 51s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a8d8193216834c5ebb41d1d8eaf6d9c8 : FAILURE in 26m 27s","accounts_in_message":[],"_revision_number":1},{"id":"806ca752540fb93be6993d46ee549e7ea0415ecc","author":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"date":"2019-09-18 18:54:19.000000000","message":"Abandoned\n\nId5e6f74a5ed96fdf17e2c241466897974aa5a218 merged, zuul checks passed","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b9a0b7d55c9757ff0eacb5ab39c474eaab1d2a6a","revisions":{"b9a0b7d55c9757ff0eacb5ab39c474eaab1d2a6a":{"kind":"REWORK","_number":1,"created":"2019-09-16 20:43:48.000000000","uploader":{"_account_id":14200,"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","username":"mmalchuk"},"ref":"refs/changes/78/682478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/78/682478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/78/682478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/78/682478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/78/682478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/78/682478/1"}}},"commit":{"parents":[{"commit":"7b56f1d9b757584b887d26160fbce66af74dddcc","subject":"Update playbook to newer syntax.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7b56f1d9b757584b887d26160fbce66af74dddcc"}]}],"author":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2019-09-16 20:43:15.000000000","tz":180},"committer":{"name":"Maksim Malchuk","email":"maksim.malchuk@gmail.com","date":"2019-09-16 20:43:15.000000000","tz":180},"subject":"[DNM] test only","message":"[DNM] test only\n\nChange-Id: Icddedc71e7ab464372e8e07fdce4c6ce79013d5b\nSigned-off-by: Maksim Malchuk \u003cmaksim.malchuk@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b9a0b7d55c9757ff0eacb5ab39c474eaab1d2a6a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/b9a0b7d55c9757ff0eacb5ab39c474eaab1d2a6a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
