)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~688453","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fpike~Id25957dac4efaa38f44012558c82e9c3dd467047","project":"openstack/openstack-ansible-os_keystone","branch":"stable/pike","topic":"goodbye_42.3","hashtags":[],"change_id":"Id25957dac4efaa38f44012558c82e9c3dd467047","subject":"Remove openSUSE Leap 42.3  jobs","status":"MERGED","created":"2019-10-14 16:48:23.000000000","updated":"2019-10-14 20:23:16.000000000","submitted":"2019-10-14 20:23:16.000000000","submitter":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688453-1571084596514-9e252c24","meta_rev_id":"7724f74cb776b31f7645305396403d1aa424115b","_number":688453,"virtual_id_number":688453,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"all":[{"value":2,"date":"2019-10-14 20:23:14.000000000","_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2019-10-14 17:26:15.000000000","permitted_voting_range":{"min":-1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-10-14 17:27:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-10-14 17:27:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 17:26:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-10-14 17:27:44.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-14 20:23:14.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"}],"messages":[{"id":"80401fd64c759e398cc9010d28dca087b7ed7710","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-14 16:48:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ad09b3b34904fef18cd439fc562b0e3815ebd616","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:26:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e3a88c712bb046eb83e55554cfa8abed : SUCCESS in 3m 39s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/1d60433d688042e28baf95b0bb41170c : SUCCESS in 4m 55s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/d0a39b8d0e2d40edb5db6d3a8843dbf5 : FAILURE in 7m 58s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/fbb78e11c03c4393a7370aaf1bea37f6 : SUCCESS in 31m 54s\n- openstack-ansible-uw_apache-centos-7-nv https://zuul.opendev.org/t/openstack/build/5d40a8d02ce644d5a78287a2f7a8b740 : FAILURE in 7m 06s (non-voting)\n- openstack-ansible-uw_apache-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/f860b01865c5412da1fbe2153197de21 : SUCCESS in 31m 01s","accounts_in_message":[],"_revision_number":1},{"id":"b3dc1802a74b2f55d21b05976ec74f39bc2c2609","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-14 17:27:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1b7c96cd444bfb36e34d130e97cc8aa7093fc59f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 18:13:51.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/e1f407a90d824231bd9d2f8d8e442050 : SUCCESS in 3m 56s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d0d6acb1b00345368d36d53f3d43e588 : SUCCESS in 6m 51s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/a05408cf3ac3445393f42453bc2da4d7 : FAILURE in 8m 26s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/ebb5ddfbef244453befefdf98f9f2993 : SUCCESS in 32m 36s\n- openstack-ansible-uw_apache-centos-7-nv https://zuul.opendev.org/t/openstack/build/228756819cd5484c8e882679d885265d : FAILURE in 7m 11s (non-voting)\n- openstack-ansible-uw_apache-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/81a4d8f487ce4ffb8f445ad5bbdbc5dd : SUCCESS in 33m 06s","accounts_in_message":[],"_revision_number":1},{"id":"5553593d6492dda16ded94498cdbf91ce82a81a6","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-14 20:23:14.000000000","message":"Patch Set 1: Verified+2\n\nThe opensuse-423 nodeset is being removed.  This change is being force-merged to correct the resulting configuration error.","accounts_in_message":[],"_revision_number":1},{"id":"f84b5465dd7723453c7a9dde9e772abeb3549a9f","author":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"date":"2019-10-14 20:23:16.000000000","message":"Change has been successfully merged by Clark Boylan","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"fb078647cb81433f8136281d5d83ce953e8fa175","revisions":{"fb078647cb81433f8136281d5d83ce953e8fa175":{"kind":"REWORK","_number":1,"created":"2019-10-14 16:48:23.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/53/688453/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/53/688453/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/53/688453/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/53/688453/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/53/688453/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/53/688453/1"}}},"commit":{"parents":[{"commit":"a026d7e5bf266f1099e38b360419492c56e60f30","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a026d7e5bf266f1099e38b360419492c56e60f30"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 16:48:01.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 16:48:20.000000000","tz":120},"subject":"Remove openSUSE Leap 42.3  jobs","message":"Remove openSUSE Leap 42.3  jobs\n\nopenSUSE Leap 42.3 is EOL since June 2019, let\u0027s remove remaining jobs so\nthat we can remove Leap 42.3 from Opendev infrastructure.\n\nChange-Id: Id25957dac4efaa38f44012558c82e9c3dd467047\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/fb078647cb81433f8136281d5d83ce953e8fa175"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/fb078647cb81433f8136281d5d83ce953e8fa175"}]},"branch":"refs/heads/stable/pike"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
