)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~688454","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fqueens~I0de9a71ae07992341d737bbbea4cdf82ab1345b1","project":"openstack/openstack-ansible-os_keystone","branch":"stable/queens","topic":"goodbye_42.3","hashtags":[],"change_id":"I0de9a71ae07992341d737bbbea4cdf82ab1345b1","subject":"Remove openSUSE Leap 42.3  jobs","status":"MERGED","created":"2019-10-14 16:49:04.000000000","updated":"2019-10-14 19:04:07.000000000","submitted":"2019-10-14 19:04:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"688454-1571079847880-9bbf562d","meta_rev_id":"33c9cf61a09d0ead432d8820419ee924101418f6","_number":688454,"virtual_id_number":688454,"owner":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-10-14 19:04:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":2,"date":"2019-10-14 17:30:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"all":[{"value":1,"date":"2019-10-14 17:30:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-10-14 17:30:57.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2019-10-14 19:04:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2723bd180ebcb2622e2abf0e8a6fce6345c40f17","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-10-14 16:49:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"550a5f0b3cef89f632d669bc2c6e13fbf1da0c61","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2019-10-14 17:30:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cfa191a5841f25a30952e66a2c0a8d594ef02349","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 17:54:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/19765c8b4bd74ba0af84af8e67446b45 : SUCCESS in 4m 50s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/42e5c96f93cd4c5a8419a941d6d8b0de : SUCCESS in 9m 05s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/88775a53f465465abcffc4489e73a55e : SUCCESS in 58m 48s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/330dafa76ca64f24810db9afa67fd8bd : SUCCESS in 31m 45s\n- openstack-ansible-uw_apache-centos-7-nv https://zuul.opendev.org/t/openstack/build/8e33bb15436e4935bbac54946cb9b83b : SUCCESS in 59m 48s (non-voting)\n- openstack-ansible-uw_apache-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/0b90e00bd4044575a133dcddc28e48f3 : SUCCESS in 37m 24s","accounts_in_message":[],"_revision_number":1},{"id":"8660577bd1f2e2ba9d853164909271b01856d76a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 18:08:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98707b92096fef74138b9932bde342f8900fc566","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 19:04:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"35036866012c22eb8dd384f014c1037dda32d663","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-10-14 19:04:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3a24721b863e4c168465542c281d26dc : SUCCESS in 4m 17s\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/97b468341d72438a978550d1c49aede9 : SUCCESS in 5m 45s\n- openstack-ansible-functional-centos-7 https://zuul.opendev.org/t/openstack/build/b7541460799d4f3bad17c980195152da : SUCCESS in 49m 36s\n- openstack-ansible-functional-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/2c61d4a401244d8cac162e08d76b22ef : SUCCESS in 35m 02s\n- openstack-ansible-uw_apache-ubuntu-xenial https://zuul.opendev.org/t/openstack/build/272debe9d627441ebf1d0253f8e90863 : SUCCESS in 35m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"97e463440c4c697e2383be58c5edcf3a1e164dd9","revisions":{"97e463440c4c697e2383be58c5edcf3a1e164dd9":{"kind":"REWORK","_number":1,"created":"2019-10-14 16:49:04.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/54/688454/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/54/688454/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/688454/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/688454/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/688454/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/688454/1"}}},"commit":{"parents":[{"commit":"30d7f6536cce7c6a150c8314df59223823741414","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/30d7f6536cce7c6a150c8314df59223823741414"}]}],"author":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 16:48:56.000000000","tz":120},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2019-10-14 16:48:56.000000000","tz":120},"subject":"Remove openSUSE Leap 42.3  jobs","message":"Remove openSUSE Leap 42.3  jobs\n\nopenSUSE Leap 42.3 is EOL since June 2019, let\u0027s remove remaining jobs so\nthat we can remove Leap 42.3 from Opendev infrastructure.\n\nChange-Id: I0de9a71ae07992341d737bbbea4cdf82ab1345b1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/97e463440c4c697e2383be58c5edcf3a1e164dd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/97e463440c4c697e2383be58c5edcf3a1e164dd9"}]},"branch":"refs/heads/stable/queens"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
