)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~693903","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Ftrain~Ic9cc341edb6f2f0ba76bd301c9782fbcc5951544","project":"openstack/openstack-ansible-os_keystone","branch":"stable/train","topic":"bugs/1847395-stable/train","hashtags":[],"change_id":"Ic9cc341edb6f2f0ba76bd301c9782fbcc5951544","subject":"Standardize on nginx-extras","status":"MERGED","created":"2019-11-12 18:41:49.000000000","updated":"2019-11-19 00:16:33.000000000","submitted":"2019-11-19 00:15:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"693903-1574122513354-8d7acffc","meta_rev_id":"81349477463ac0a53a765af6b1ffd214ac89db16","_number":693903,"virtual_id_number":693903,"owner":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-19 00:15:13.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"all":[{"value":1,"date":"2019-11-18 20:30:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":2,"date":"2019-11-18 15:45:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-11-18 21:49:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-18 21:49:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"value":0,"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-18 15:45:45.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2019-11-18 20:30:28.000000000","updated_by":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"reviewer":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"state":"REVIEWER"},{"updated":"2019-11-18 21:49:04.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2019-11-19 00:15:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"63fc60ecdc31709ca4826ba111c9ed7c61422b06","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2019-11-12 18:41:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"82684fec4eed68abc23517e62fa06de31574ff65","author":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"date":"2019-11-12 19:25:06.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"0184ab86a9cc7bcd14e1478d2c54aafb176cd9e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-12 21:08:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/e80f003699f546209a8900f094bc56a1 : SUCCESS in 5m 42s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/be9673ec42054722a6377a9ecab998f8 : SUCCESS in 1h 20m 25s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/15d8560fc4aa45319da351a492b90e44 : SUCCESS in 1h 37m 51s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3f197f9930a948f496249a6a75a5525c : FAILURE in 23m 20s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/23a4a297cf9d4fa79829521fbf1b3448 : SUCCESS in 1h 11m 00s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5b48ace69d944b19848f9eb43fa0d761 : SUCCESS in 2h 21m 07s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/649fbe2763e94301ad2c4451c60b9433 : SUCCESS in 1h 50m 26s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/653ceb0e001a42389c7b83186075ea1c : SUCCESS in 1h 02m 27s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/71f14ef695254da49b6b9ae5c4ae190b : FAILURE in 20m 28s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e150897c125243be80723d8229a62013 : SUCCESS in 46m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc7ddee69f6841a48e3c353db048aee7 : SUCCESS in 4m 44s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/e578b09a653f4c5bbed6315e6c4e9ffa : SUCCESS in 37m 57s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d37266d3813641488d014ebf9d216543 : SUCCESS in 30m 02s","accounts_in_message":[],"_revision_number":1},{"id":"e97a03f9301001d8c87f5fe339a36fbb24b7240d","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2019-11-13 05:35:53.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"01ba5f46713cd5a2cbd8e403468e413c676678cb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 08:43:13.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2fdd10ceb96e4bcfafee614b7e07d5dd : SUCCESS in 6m 14s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/5a03ce09c343499898c179af5e31b6a9 : SUCCESS in 1h 20m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2e78cdbd2e8140c48f9ccd412ab6201e : SUCCESS in 1h 42m 34s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0ee14c1d068c453c8ac69cd3f5e44bee : FAILURE in 23m 36s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b08359fddd044fa9891c86e43f19125 : SUCCESS in 1h 18m 44s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f86c3c585fab4356bbc66300eed664ab : TIMED_OUT in 3h 05m 22s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1a576d64f2d342e5ba77ed5cc2c4a574 : SUCCESS in 1h 52m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9bc1fe59262740f8ab4bf0d5aed49eb6 : SUCCESS in 1h 16m 17s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/1b299314abaa42daa73d4fe6689bacff : FAILURE in 22m 05s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9de9fbe282d4947845d4581c414c2ef : SUCCESS in 55m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f1d3df0a84834af1836a979d1460e401 : SUCCESS in 5m 16s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/a08f4cc9b58d46bd8412621e8d347cd6 : SUCCESS in 44m 21s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/59bf1e9f35154b02b78da93139402e82 : SUCCESS in 34m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f805215cf36ab2793678ef2480a9518f2c710612","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2019-11-13 18:46:33.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4fe5ea7192a5084b74f72f4c9e801f49ecbc4118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-13 21:17:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/beceb9b761834c2cb9851131fd7f69eb : SUCCESS in 7m 25s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/8f7a083da3af40a5b3386a92b9853213 : SUCCESS in 1h 12m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2a48a5d50d0648f79fb4c8441b4775ea : SUCCESS in 1h 43m 17s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/a91843faa2a9461e9df95db827bf5d46 : FAILURE in 27m 40s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a6b05eb7d9364fe9ac98f0df7a0d415c : SUCCESS in 1h 01m 42s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b6e0158b51d34cb58d5192a8750f67e4 : SUCCESS in 2h 14m 42s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/64504e5136f5420199e476f3ff352795 : SUCCESS in 1h 50m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0d4321c7f8a040e4bb4f3b6ff2133041 : SUCCESS in 1h 09m 23s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/eee0dcd9b682493d96158c2887a026dc : FAILURE in 20m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3ba0052c74444120b1cd95e210704c40 : SUCCESS in 1h 00m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/629a08d8a20041269525a3d8c8947d86 : SUCCESS in 5m 02s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/48887302373c4767a385eee1c122cf3c : SUCCESS in 39m 48s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/393afa1cd0244167bc2b8560593bea94 : SUCCESS in 34m 59s","accounts_in_message":[],"_revision_number":1},{"id":"b9cf5c27949483de6d986598c8e08b06376009ad","author":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"date":"2019-11-15 16:45:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4d1d24dfe72e59c56ff3deb2a7a2700302e4420a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-15 19:22:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ac1263ca5af64b2888ff538665191696 : SUCCESS in 5m 46s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/227835f319674485b4226fce726c79ba : SUCCESS in 1h 13m 18s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/23f1f5231676436083782b053b04d8b2 : SUCCESS in 1h 48m 38s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/13f9b8589e5f485587127beb0dda6771 : FAILURE in 24m 18s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f73482716a474d0186e74d9f52c336ec : SUCCESS in 1h 19m 26s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e4a55c9033474d65ae0982b416f65a06 : SUCCESS in 2h 30m 56s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7be87a8e8c641f1912d48b4815a776d : SUCCESS in 1h 41m 09s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/759fc23c9d0c473f833fa60e9b6abf90 : SUCCESS in 1h 05m 49s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8f55c4ce2dc84f65864b72560328e0fb : FAILURE in 18m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/376946f495e24f9387af4ba08d03fe85 : SUCCESS in 49m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55aef5b80ef941d2909a9d665f26a17b : SUCCESS in 4m 28s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/f78ddde932794fff92ebc84ef2299c27 : SUCCESS in 44m 02s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3b5f4197cf0045638663ce14bbf78194 : SUCCESS in 28m 40s","accounts_in_message":[],"_revision_number":1},{"id":"20a97b4bfc2f98a0500702e8d266826d67e16703","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 10:03:30.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"aa47c39b767327ac910744d22b1004da408bbe63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 12:36:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ae486a7f9fb14ff4aca43b4568da14d7 : SUCCESS in 8m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/ac0288d1b86540b3a0915e5c8fdda766 : SUCCESS in 1h 21m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/25f1c12a60284cd3a546ec065980b5f8 : SUCCESS in 1h 36m 57s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9664250a2f824f28858adaefe855bd81 : FAILURE in 29m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fddc864d3f1843cc88c59f5d54d96e20 : SUCCESS in 1h 22m 13s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e8f96664c12a49c09ecb89feb6ccd85f : SUCCESS in 2h 28m 23s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d58823c28c72440da373548bad215fd5 : SUCCESS in 1h 52m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a71ce7f978f24a759bf80a90f1a05ede : SUCCESS in 1h 08m 23s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/9a6f3152d328415681bcdac5e931c03c : FAILURE in 21m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3759bf59e44c49cc89c248864ce0cc50 : SUCCESS in 49m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf6e4b4c300c4077888247df069e2f4e : SUCCESS in 4m 17s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/0206028f0e284b51b6a7e43b62a3254e : SUCCESS in 38m 51s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0e9ca3819c3347f09f98f753b1251754 : SUCCESS in 32m 05s","accounts_in_message":[],"_revision_number":1},{"id":"3f44641a6de20e7976f65ce20491feb9b0564cc0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 14:54:12.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"f68383032e1ab5e38b9508b1439376bba18400bf","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2019-11-18 15:45:45.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c1219c715adb3ab48f53bacdd58db462f7c12d7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 17:33:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/97353b7b531145af9dc815d86a40a572 : SUCCESS in 6m 48s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/945d3dadf3014a73b93f80bb4ebc9a66 : SUCCESS in 1h 24m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/aee953d527fe458882b3491a7170cb96 : SUCCESS in 1h 27m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fd965c0843bc4f449c538cd7f84188e5 : SUCCESS in 1h 33m 22s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97df1a8579a94d9d821aebfa3b4b940f : SUCCESS in 1h 29m 56s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5308ea4dc42e40468721a474cbff412f : SUCCESS in 2h 37m 28s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/58fec028859244e9b77675df5206645e : SUCCESS in 1h 54m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d77716e84ef04e3fa5e7852b87ed3412 : SUCCESS in 1h 09m 23s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f4365d53490f46dea541903342208ea7 : SUCCESS in 1h 05m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/65d0330719724fb2af231fb8d11c40ea : SUCCESS in 1h 05m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4402a73f05448f0ad956c993e7330f4 : SUCCESS in 7m 00s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/ce00f3e6bb6940469e66f1e6301e19e8 : SUCCESS in 52m 01s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/649f3d614d344f8694f48731cef09288 : SUCCESS in 46m 03s","accounts_in_message":[],"_revision_number":2},{"id":"bab296b972f254d050e256b767d4d79703b29d70","author":{"_account_id":21314,"name":"Cam J. Loader","display_name":"Cam J. Loader","email":"cameron@loaderusa.net","username":"4-cameron"},"date":"2019-11-18 20:30:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8293f07d302b20e7213d559b78906d24d847f68b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2019-11-18 21:49:04.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"d692e83e2995cefb80d07f6fef8d8cb3566b494a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-18 21:49:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1650976b90d6236e2de7ff77b7fd760157872642","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 00:15:13.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/08492f57e7d74a0985b84d0684bd5d32 : SUCCESS in 5m 08s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/887ae46df61f4bdf9de3f2977e477800 : SUCCESS in 1h 13m 58s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/10bfd41473814e1883d3f7e12f42ddb0 : SUCCESS in 1h 20m 08s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c77471905e6d4e5db695ca54b09c6e79 : SUCCESS in 1h 29m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/337be832de584b549f27412bff812126 : SUCCESS in 1h 09m 54s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c387a2ec9cb340a3a3c5eca9b79da51a : SUCCESS in 2h 22m 05s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c0070b22987e4a6f8cd660285cbb2b8e : SUCCESS in 1h 38m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6b69acc3ec2743fb93ecae62699734b5 : SUCCESS in 59m 23s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b9610121232843cbabdc4ac2457796ad : SUCCESS in 47m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/080fb30fed90469293ab532706c820b9 : SUCCESS in 4m 45s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/0767e700d797473f817ca6c1c2c17946 : SUCCESS in 38m 15s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a795c2f1d824e1fbe5b628a32cdc4bb : SUCCESS in 29m 15s","accounts_in_message":[],"_revision_number":2},{"id":"1ee8fe84ac32d98632664afc2f93efcddc9db435","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 00:15:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b585f22d79ba6d273d3e2dfbf9259f7fff95efbe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-19 00:16:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdd0d7a455a349fc880b0db788497e5f : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bcca76a29eedcd7608c2d1e0e460b9850bdb00cf","revisions":{"ac58714b5cea5319a5f7ff472fa74662a7762c63":{"kind":"REWORK","_number":1,"created":"2019-11-12 18:41:49.000000000","uploader":{"_account_id":7414,"name":"David Wilde","email":"dwilde@redhat.com","username":"d34dh0r53"},"ref":"refs/changes/03/693903/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/03/693903/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/693903/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/693903/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/693903/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/693903/1"}}},"commit":{"parents":[{"commit":"ddd55f0f1e40d14369579c1a415b4118a946959e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ddd55f0f1e40d14369579c1a415b4118a946959e"}]}],"author":{"name":"Dave Wilde (d34dh0r453)","email":"david.wilde@rackspace.com","date":"2019-11-05 15:42:22.000000000","tz":-360},"committer":{"name":"Dave Wilde (d34dh0r53)","email":"david.wilde@rackspace.com","date":"2019-11-12 18:41:49.000000000","tz":0},"subject":"Standardize on nginx-extras","message":"Standardize on nginx-extras\n\nThe use of nginx-full causes a service restart on package update\nwhich brings down the keystone endpoints.\n\nChange-Id: Ic9cc341edb6f2f0ba76bd301c9782fbcc5951544\nRelated-Bug: 1847395\n(cherry picked from commit 461245350b52d9ba6edc0ac474c6c6d2cb16ea3e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ac58714b5cea5319a5f7ff472fa74662a7762c63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ac58714b5cea5319a5f7ff472fa74662a7762c63"}]},"branch":"refs/heads/stable/train"},"bcca76a29eedcd7608c2d1e0e460b9850bdb00cf":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-11-18 14:54:12.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/03/693903/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/03/693903/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/693903/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/693903/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/693903/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/03/693903/2"}}},"commit":{"parents":[{"commit":"ddd55f0f1e40d14369579c1a415b4118a946959e","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ddd55f0f1e40d14369579c1a415b4118a946959e"}]}],"author":{"name":"Dave Wilde (d34dh0r453)","email":"david.wilde@rackspace.com","date":"2019-11-05 15:42:22.000000000","tz":-360},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2019-11-18 14:54:10.000000000","tz":0},"subject":"Standardize on nginx-extras","message":"Standardize on nginx-extras\n\nThe use of nginx-full causes a service restart on package update\nwhich brings down the keystone endpoints.\n\nDepends-On: https://review.opendev.org/694759/\nChange-Id: Ic9cc341edb6f2f0ba76bd301c9782fbcc5951544\nRelated-Bug: 1847395\n(cherry picked from commit 461245350b52d9ba6edc0ac474c6c6d2cb16ea3e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/bcca76a29eedcd7608c2d1e0e460b9850bdb00cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/bcca76a29eedcd7608c2d1e0e460b9850bdb00cf"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
