)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~702714","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I16a61501a0578e87464e8bd1031af4270eac2d98","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"bug/1856165","hashtags":[],"change_id":"I16a61501a0578e87464e8bd1031af4270eac2d98","subject":"Delegate identity provider setup to service setup host","status":"MERGED","created":"2020-01-15 18:17:06.000000000","updated":"2020-02-17 11:20:33.000000000","submitted":"2020-01-16 20:15:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702714-1579205717027-1b68ee4e","meta_rev_id":"f397a8b237651137a0638de86147cf09b3c64964","_number":702714,"virtual_id_number":702714,"owner":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 20:15:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-01-16 14:32:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-16 18:20:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-16 18:20:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-16 14:32:24.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-01-16 18:20:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-01-16 20:15:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d17127d287609da7caad1e0043adad2690eb6eb","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-01-15 18:17:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20ad0b47019383c7b074398a615bd032b7e8b3d3","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-01-15 18:18:11.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"71e53f6b04f7dc6fe9a75c5030841f1aa10da9e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-15 21:19:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9ecfdee3b95b4b4c889a5b055ecdf1fd : SUCCESS in 10m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d9b1498aaa55402c862eff5058cf45e2 : SUCCESS in 1h 05m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/187220a9f9f94688bf9cd45a3d6d7b90 : SUCCESS in 1h 36m 11s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fc16d674657f45b8ab88814b9e772834 : SUCCESS in 1h 29m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12067c59989641ad9e8e6764341107d7 : SUCCESS in 1h 07m 20s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d8ede527a6a74d7184035ae187ba75c1 : SUCCESS in 2h 52m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/08024687be9f43d4a16fb504ae1c8c06 : SUCCESS in 1h 50m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/184b7831262147e3bddac0b0ddab8fac : SUCCESS in 1h 10m 27s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/ce9756a10b3b4852bcecd09f4d5622ae : SUCCESS in 1h 01m 59s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0021563df468497eaf7ffcca6a9a484c : SUCCESS in 59m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9cf0b04c6fd4d5db87f1d3688fe755d : SUCCESS in 9m 37s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/54f5a133079349eba3e7b68a20b54c17 : SUCCESS in 44m 57s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/320954834ce14e968d054ace96751acc : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":2},{"id":"10983bfdb19ec10b53ebbb7f3884b234d6a30434","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-01-16 10:35:21.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ca7558b7ee9d58e0070b710e216d486520980572","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 13:07:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/950ddb3a231a43a989d3270593939419 : SUCCESS in 6m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1dffd668a35049c985c20b558119e1e2 : SUCCESS in 1h 20m 30s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1e8f2d65098b4343a48079574cf0ef6f : SUCCESS in 1h 31m 28s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/29f9d195ee1347b99cb07d09e5b8f12f : SUCCESS in 1h 35m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4cefcc20910c43c89839545a90b176ac : SUCCESS in 1h 27m 19s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3b5c3996811749da88d377276cb71ba5 : SUCCESS in 2h 30m 41s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c1338d8854004ebdbd7812eeb676efc7 : SUCCESS in 2h 02m 53s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2c04d902b21644c2a60704cc68129391 : SUCCESS in 1h 06m 10s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f6c09e37ca4045ee844f33ffe8247eb2 : SUCCESS in 1h 14m 49s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3220613cd6294e35abb62001535e28b9 : SUCCESS in 51m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9bfc4448b024bff9414bae0f781c38f : SUCCESS in 6m 43s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/131df32c2ee0441ba59110a1bda2da72 : SUCCESS in 47m 28s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ffe2cebf18824398b769d80700170420 : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":3},{"id":"eb3037ae289ae636ec72fae93b282b10c3813cd5","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-01-16 14:32:24.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"9aed5ca4b0d68d797dd928c53f0a8266f176f15c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-01-16 18:20:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"6350156337212990a750154795e46ab7a5c12906","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 18:21:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"398db2fcbbeb93bb9737f45ab8b9c8b2a91836bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 20:15:16.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c2555352d0764a6684cbd0238e6f0d79 : SUCCESS in 6m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bbadb5b209fa48b0b96fedbb98d96ffb : SUCCESS in 1h 09m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fcb5e7bb1ab44ed29e57d28cfdeb3f0b : SUCCESS in 1h 37m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/dd6a397b3f20412c98431a338a600fa8 : SUCCESS in 1h 31m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/77e020e3c1d84523bf41327b47d5d2e8 : SUCCESS in 1h 38m 51s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7da0329dbb194232a7c903e78e293bf1 : SUCCESS in 1h 46m 51s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3d6eee24a9374ee9a16a8c562e1954ea : SUCCESS in 1h 04m 56s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/599dd0131c6f470c9990421a104497e8 : SUCCESS in 55m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b45dbab56b8d41038b7fd268a2d995a5 : SUCCESS in 7m 06s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/e3e7700fa24c431382d0c03af9342b57 : SUCCESS in 34m 40s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6c2fe4bd7c8e4bcea908e199023052fb : SUCCESS in 34m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a046b1fa29c4e8894561b0254dfced61d0f22251","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 20:15:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"28f5d15998bd844b74f528d60bb80481ed4615e8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 20:16:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9085027dc7d349858c9e704639aa93df : SUCCESS in 1m 13s","accounts_in_message":[],"_revision_number":3},{"id":"778b51f8a3e2eec8fb852b21bbb6cb26b7e62e4f","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-02-17 11:20:33.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit e6e445a9acbee6c83314305298d635f242fe8dc0","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8a31f80e35a4ebad83da260679952a823bf9fbf9","revisions":{"03704eb63c0db85fb6b57832d9cb8c8a5e22837d":{"kind":"REWORK","_number":1,"created":"2020-01-15 18:17:06.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/14/702714/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/702714/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/702714/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/702714/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/702714/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/702714/1"}}},"commit":{"parents":[{"commit":"2eb0af8f8b94afa0703194d7ffe1e5e39977b7a5","subject":"Merge \"Replace Module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2eb0af8f8b94afa0703194d7ffe1e5e39977b7a5"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-01-15 17:55:58.000000000","tz":0},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-01-15 17:55:58.000000000","tz":0},"subject":"Openstacksdk is not available in all installs","message":"Openstacksdk is not available in all installs\n\nDelegate to keystone_service_setup_host to enable tests to pass for situations\nwhere the openstacksdk installed on the keystone container host.\nOtherwise os_ tasks will fail with: openstacksdk is required for this module\n\nThere has been a bug logged for the error that would happen prior to this\npiece of code being moved to the ansible os_ modules in which would complain\nabout python-keystoneclient not being on the keystone container host due to\nthe same deployment circumstances.\n\nChange-Id: I16a61501a0578e87464e8bd1031af4270eac2d98\nCloses-Bug: #1856165\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/03704eb63c0db85fb6b57832d9cb8c8a5e22837d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/03704eb63c0db85fb6b57832d9cb8c8a5e22837d"}]},"branch":"refs/heads/master"},"19fba8015b431c863ddd31db74c577bc94d6d5f1":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-01-15 18:18:11.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/14/702714/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/702714/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/702714/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/702714/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/702714/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/702714/2"}}},"commit":{"parents":[{"commit":"2eb0af8f8b94afa0703194d7ffe1e5e39977b7a5","subject":"Merge \"Replace Module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2eb0af8f8b94afa0703194d7ffe1e5e39977b7a5"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-01-15 17:55:58.000000000","tz":0},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-01-15 18:18:05.000000000","tz":0},"subject":"Openstacksdk is not available in all installs","message":"Openstacksdk is not available in all installs\n\nDelegate to keystone_service_setup_host to enable tests to pass for situations\nwhere the openstacksdk is not installed on the keystone container host.\nOtherwise os_ tasks will fail with: openstacksdk is required for this module\n\nThere has been a bug logged for the error that would happen prior to this\npiece of code being moved to the ansible os_ modules in which would complain\nabout python-keystoneclient not being on the keystone container host due to\nthe same deployment circumstances.\n\nChange-Id: I16a61501a0578e87464e8bd1031af4270eac2d98\nCloses-Bug: #1856165\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/19fba8015b431c863ddd31db74c577bc94d6d5f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/19fba8015b431c863ddd31db74c577bc94d6d5f1"}]},"branch":"refs/heads/master"},"8a31f80e35a4ebad83da260679952a823bf9fbf9":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-01-16 10:35:21.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/14/702714/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/14/702714/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/702714/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/702714/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/702714/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/14/702714/3"}}},"commit":{"parents":[{"commit":"2eb0af8f8b94afa0703194d7ffe1e5e39977b7a5","subject":"Merge \"Replace Module\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2eb0af8f8b94afa0703194d7ffe1e5e39977b7a5"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-01-15 17:55:58.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-01-16 10:35:19.000000000","tz":0},"subject":"Delegate identity provider setup to service setup host","message":"Delegate identity provider setup to service setup host\n\nThis patch delegates tasks setting up a keystone identity provider to the\nkeystone_service_setup_host becasue the ansible modules used require the\npython-keystoneclient module. These tasks will fail when the keystone\n(container) host is not the deploy host as python-keystoneclient will\nbe missing.\n\nThese tasks have recently been refactored in a previous patch [1] to use\nthe os_* ansible modules instead of the OSA specific keystone module.\nThis patch fixes a reported bug with the old code which is also present\nafter that refactoring.\n\n[1] https://review.opendev.org/#/c/656397/\n\nChange-Id: I16a61501a0578e87464e8bd1031af4270eac2d98\nCloses-Bug: #1856165\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8a31f80e35a4ebad83da260679952a823bf9fbf9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8a31f80e35a4ebad83da260679952a823bf9fbf9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
