)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~702920","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I5ce4d0ba4c1522fd3899fc97e4d0b6f064b47bb4","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"bug/1856165","hashtags":[],"change_id":"I5ce4d0ba4c1522fd3899fc97e4d0b6f064b47bb4","subject":"Fix assurances for federation scenario","status":"MERGED","created":"2020-01-16 16:35:48.000000000","updated":"2020-02-21 10:17:43.000000000","submitted":"2020-01-18 10:32:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"702920-1579343578611-bbef5f2b","meta_rev_id":"bde93481a47c0d143db66aa2c0decd36e5e9a6db","_number":702920,"virtual_id_number":702920,"owner":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-18 10:32:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-01-17 16:13:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-01-18 08:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-01-18 08:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-01-17 16:13:18.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-01-18 08:45:44.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-01-18 10:32:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"87493685cc75421f00f0ebd3baa34e457e69741e","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-01-16 16:35:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d38e6fc691ed055265bb8412ee2980d6e2ba83c3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-01-16 16:38:37.000000000","message":"Patch Set 1: Code-Review+1\n\nIt doesn\u0027t need depends-on since it\u0027s rebased on top of that patch. Otherwise lgtm","accounts_in_message":[],"_revision_number":1},{"id":"e3231db421cd17164418a036a63a99b4b045790b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 19:40:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9a542c2c05644eeb83a4aecbb75dc76e : SUCCESS in 8m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/726251825deb4a0fbc5c362ddcc9b308 : SUCCESS in 1h 05m 09s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f627e2adb734420db36fec2d07ba4a30 : SUCCESS in 1h 35m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b259e751d930430dbb9373ac90aa8d0a : SUCCESS in 1h 37m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aa4d321a58764b8e99ca24be37c51ae2 : SUCCESS in 1h 11m 03s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2834dcfba5cc4507a1cd68584b65a962 : SUCCESS in 2h 57m 43s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/44c97a63bd6f4313a06c0c7230f0fca9 : SUCCESS in 2h 03m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ebb53b3efec3437698420bdec9f4b8b9 : SUCCESS in 1h 20m 50s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fef7e4c56fa44b5a94262c69b9286912 : SUCCESS in 59m 24s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b529912f38a8410aa96c12d188954c64 : SUCCESS in 52m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc40e46d8f984347aea75b13edb1eba2 : SUCCESS in 5m 39s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/6fb6260947494a38bbb21c3abe361125 : SUCCESS in 41m 16s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f1cbfb7bf27c41b5af9d1e23307ce0d7 : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":1},{"id":"6a909e9c246fa723b9795596d512782a88361b40","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-01-17 16:13:18.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b89b57eabe1d5951ddc32a868c5cf789f90cb936","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-01-18 08:45:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f4960b3bbe04be38782e2e8ed04961c5b75566a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 08:45:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"95fef813d803508c8a8650c92beb9bafff7c1351","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 10:32:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7a3f10146f9469fdd16f2bb916c3c085ad176a9a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 10:32:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0b55d58f9f93418e9eccfc843edadcc9 : SUCCESS in 5m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/84f30bb4004c408ea1ed07348d8b9908 : SUCCESS in 1h 03m 05s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3c229cb4cbc94fa2a0e17e71e9d82eac : SUCCESS in 1h 38m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3e28925527a049c8b095501a46256e72 : SUCCESS in 1h 16m 42s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd94b39afbf54c4cb33705de9a712156 : SUCCESS in 1h 07m 15s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a2eebf0eee2d473095523d6a1134c5fb : SUCCESS in 1h 42m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7bc374e843254de8b073654f37d30b78 : SUCCESS in 1h 04m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d25590aed9f240d6913075261f9209b2 : SUCCESS in 50m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80200a1bb1754f66bf36cbd78b2a9a2f : SUCCESS in 6m 51s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/ef61a84cf89947d289f35cdf12c1c67a : SUCCESS in 54m 28s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d77d9669af348d6bf89af40da239578 : SUCCESS in 29m 47s","accounts_in_message":[],"_revision_number":1},{"id":"ee809f523e6d323f96a204fc76c3b7d93b2adc68","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-18 10:34:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebb2f63a02f3458e8c534827d72cb2a5 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1},{"id":"a8ffa1efd9ba73855e149ae0cc2127d4f1805919","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-02-21 10:17:43.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 9ade2e440a98f9e7d7dabf867da0f01fcaefa8c2","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e68231cc610b07593f2b832c3733d90f9d0bb06e","revisions":{"e68231cc610b07593f2b832c3733d90f9d0bb06e":{"kind":"REWORK","_number":1,"created":"2020-01-16 16:35:48.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/20/702920/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/20/702920/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/20/702920/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/20/702920/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/20/702920/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/20/702920/1"}}},"commit":{"parents":[{"commit":"8a31f80e35a4ebad83da260679952a823bf9fbf9","subject":"Delegate identity provider setup to service setup host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8a31f80e35a4ebad83da260679952a823bf9fbf9"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-01-16 16:33:46.000000000","tz":0},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-01-16 16:33:46.000000000","tz":0},"subject":"Fix assurances for federation scenario","message":"Fix assurances for federation scenario\n\nTested the move to the os_* modules, couple of issues needed to be fixed.\n\nDepends-on: https://review.opendev.org/#/c/702714/\nChange-Id: I5ce4d0ba4c1522fd3899fc97e4d0b6f064b47bb4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e68231cc610b07593f2b832c3733d90f9d0bb06e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e68231cc610b07593f2b832c3733d90f9d0bb06e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
