)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~709493","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I0c197ec41eac4605e5c87cbeed59730d766bb5da","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I0c197ec41eac4605e5c87cbeed59730d766bb5da","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-02-24 13:11:06.000000000","updated":"2020-02-25 13:49:22.000000000","submitted":"2020-02-25 13:46:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"709493-1582638401225-13a694a1","meta_rev_id":"ac960e4b3501ea34ce814a69689b17987f2c2d4f","_number":709493,"virtual_id_number":709493,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-25 13:46:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-25 11:56:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-25 11:56:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-25 11:56:58.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-02-25 13:46:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a6a8ecb7283c80c4a2d57252a9221f82491d3c0e","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-02-24 13:11:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"463078517b8458095f6a09d764974888725eb66b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-24 14:57:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d5f3d22883ac45a68b7481225295951a : SUCCESS in 8m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/762d807ece014133a5a9c26b5a14579c : SUCCESS in 1h 04m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2c832f7b903941549011ca1a2bab5f80 : SUCCESS in 1h 28m 26s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/28bdd8e86df6465480067cc513929ba9 : FAILURE in 55m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d1a1bb9af0ff41989eee5b0161993d00 : SUCCESS in 1h 15m 11s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2f9d21e29c6d4324aeccb903bf80d6b0 : FAILURE in 1h 30m 19s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fddc3bfaa94b4caca2513393d7b60123 : FAILURE in 1h 00m 58s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/195aeb3d46774419828e8d8242d909e6 : SUCCESS in 1h 15m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e8e595a92e6a4533926423ea8f48e8f3 : FAILURE in 31m 20s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a91579de8a654ba7a94ed3b5defc9fc7 : SUCCESS in 1h 00m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aee906bcc8894bb0a76124b51282ce5f : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6934a7385d3a499b931ad0aa2cfad540 : SUCCESS in 6m 04s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/7b624aec85c94d9eac80e954f3689e8b : SUCCESS in 38m 51s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53758d782e1f4a968675144a6b58cbb7 : SUCCESS in 33m 31s","accounts_in_message":[],"_revision_number":1},{"id":"464b7870774b6b5ae6d70e47d70f0210d1dbc34b","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 08:52:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d572a25b77bbafe3c168a9f6443b7f667ce38e18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 10:50:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b985e18c229c43d7924628a3a694583a : SUCCESS in 10m 58s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/66df59ca24ac4cf1ab543280b1ba6964 : SUCCESS in 1h 09m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c53e888cf05d4e9d9a67bfaab3508211 : SUCCESS in 1h 35m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/65a2e598edb242a5a16c125fb70693e4 : SUCCESS in 1h 23m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3abfc046e5ab498c8421e03cae3cc9ec : SUCCESS in 1h 09m 13s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/56ac0750ffda49adbed08b90ffad3777 : FAILURE in 1h 24m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54c07cf85abb47d6a2f1cdeefea08434 : FAILURE in 1h 07m 32s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/166b15ea725e4abcb86dc01fb598060d : SUCCESS in 1h 09m 40s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2c5b9827de674a9da2999e7ffc8a3dac : FAILURE in 29m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7932cdfce6c14a8d93c6b235f613ad43 : SUCCESS in 56m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0441cb01ca1a4c1bad720e564235d90c : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/253ab4c6f165431b99eb09559595517e : SUCCESS in 5m 15s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/4c2a4a7e5a194305b261b039dd744a9c : SUCCESS in 42m 58s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/523b4342ab1a4905a0ea6641028cb815 : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":1},{"id":"e7136e3ceb80f6f3eddbd34edeb5397ada2137da","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-02-25 11:56:58.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"97913d501b7fb85a1e3c20f0beac8fa7ad4bbb9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 11:57:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2b1b6b194be5774f14c7f6db0c52d57e3ffb5590","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:46:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/155d0f1574374560b67476c365a07c32 : SUCCESS in 8m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5545ec8f3527413f8c4db16c2faf55f0 : SUCCESS in 1h 06m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/70090eb5b39144cbbc84000930c5699b : SUCCESS in 1h 41m 24s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fbd719bb7bb447cf9b552d528dcd983a : SUCCESS in 1h 29m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/938cf60f201240e2a696043f2aee204d : SUCCESS in 1h 24m 21s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a2e198f2242f4435b2c525c27c4f61b0 : SUCCESS in 1h 19m 28s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3f030dc293d34bd89f286a93916a667d : SUCCESS in 1h 01m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5818f8da1e0c46b1b5bc7c92a49ff18c : SUCCESS in 8m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4a0993bf7dc44e9936c32aaf7963c69 : SUCCESS in 7m 26s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/2932378a4415437395cff1420e13522b : SUCCESS in 40m 38s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1852f2ddd4264ceca00c985ef0bec2ac : SUCCESS in 40m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7cce93ec2b4fe5fe916a5a78a3eda253d067977e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:46:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6b39e9770e8a6f8a3922d30b8620f8d0ccfcd42d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-25 13:49:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb73fdd617df4a89be7ff039f353f6aa : SUCCESS in 2m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/07db340bacc14728934bd3ce638e65be : SUCCESS in 2m 17s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"249bf83c753e0726f5ba23935b15bb224acbf6a2","revisions":{"249bf83c753e0726f5ba23935b15bb224acbf6a2":{"kind":"REWORK","_number":1,"created":"2020-02-24 13:11:06.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/93/709493/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/93/709493/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/709493/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/709493/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/709493/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/93/709493/1"}}},"commit":{"parents":[{"commit":"41e75786ae04ed2f46a4cae2918a1069577915a4","subject":"Merge \"Fix federation scenario assurances os_user usage\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/41e75786ae04ed2f46a4cae2918a1069577915a4"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:11:02.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-02-24 13:11:02.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I0c197ec41eac4605e5c87cbeed59730d766bb5da\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/249bf83c753e0726f5ba23935b15bb224acbf6a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/249bf83c753e0726f5ba23935b15bb224acbf6a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
