)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~711429","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ib34fdb680ecdda6c50be995af2c980aa274f2ba2","project":"openstack/openstack-ansible-os_keystone","branch":"master","hashtags":[],"change_id":"Ib34fdb680ecdda6c50be995af2c980aa274f2ba2","subject":"Change from python-memcached to pymemcache","status":"MERGED","created":"2020-03-05 10:47:09.000000000","updated":"2020-03-13 23:45:51.000000000","submitted":"2020-03-13 23:44:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"711429-1584143057825-46c24852","meta_rev_id":"515931aecf7fa5b71147f6dbe85f0161bc178122","_number":711429,"virtual_id_number":711429,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-13 23:44:17.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-03-06 06:58:33.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":2,"date":"2020-03-13 22:04:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":2,"date":"2020-03-05 13:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"all":[{"value":1,"date":"2020-03-13 22:04:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-05 13:15:57.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-03-13 22:04:10.000000000","updated_by":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"reviewer":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"state":"REVIEWER"},{"updated":"2020-03-13 23:44:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82a7984c1f0b5072df482e4f4d850c45e2b18709","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-05 10:47:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e80272062c5ee826c806fbf0369319d1f654834b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 12:06:47.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/648f82668a264a0e9f3033750b35ee23 : SUCCESS in 7m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a1995435114f4ee2b67f0f3550d0d354 : FAILURE in 27m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0ddd9649099a49bd8e69bd83d1f677ba : FAILURE in 40m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7415b6bff69b4660895072129b365663 : FAILURE in 32m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbfec4ed2a104b0c93c84edad1fc14ec : FAILURE in 25m 03s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cd13b5e4e08f4f10a85effae34c069e4 : FAILURE in 1h 16m 36s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d099bb3818bc4f858be3cd90ca901db2 : FAILURE in 1h 00m 44s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/104e4fcbec6140cab9e17d23d584f591 : SUCCESS in 1h 04m 42s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/02e6b23b24db4da5a71cde393beaeb75 : FAILURE in 31m 01s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f762bd2ce1684e0abb0a7a0ba61f28ec : SUCCESS in 56m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30082c4d78524c36a704b88b7785fa28 : SUCCESS in 5m 21s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/1e23f91dfa1a47ff94034cf8d9475a79 : FAILURE in 28m 51s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8e7176c686b8450d8e751b9c5d089643 : FAILURE in 26m 15s","accounts_in_message":[],"_revision_number":1},{"id":"621bf436a35865b9d37add873be0cb85cc43001f","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-05 12:41:59.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c5033e5c3f4ddd3e744654d4251b37d7f3b9a970","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-05 12:42:24.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a6d8b6a5bb4198ec579babd3ac164d67eafb6804","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-05 13:15:57.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e806b010ca39ddd9a71a08f55b659f308d304485","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 14:12:22.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/25f666bb047249cd83ff0358d6c3d730 : SUCCESS in 7m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ace90202de2f4a8e87e2020e75a39c70 : SUCCESS in 1h 02m 27s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0867fd05dbf0430394be1a2272942c65 : SUCCESS in 1h 25m 43s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3eda54f2844f4009a2283d9d8548f34b : SUCCESS in 1h 23m 39s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/50234b67835b47b083422495c20e395a : SUCCESS in 1h 21m 24s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/982bebc7e25e4203af830744ec61d858 : FAILURE in 1h 23m 48s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a83d6aa33ef14699ababb9fe7ff9ed2b : FAILURE in 1h 10m 06s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fe0acf511978493a91d9716e948fa5bd : SUCCESS in 1h 17m 25s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/41036f1105954c1caf9570b2c734d95f : FAILURE in 30m 03s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/001bc21f7b8540d6a358e1f9e741dbec : SUCCESS in 57m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b29115d2d31945f19fb76119877c20f0 : SUCCESS in 5m 54s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/6955e2c039f147d686a37e71b62c731f : FAILURE in 31m 56s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e387b5e0970f47ef899e3a83b2ea8e7e : FAILURE in 25m 43s","accounts_in_message":[],"_revision_number":3},{"id":"a2d8a2bbb9bfd9d64730e37e32ff7e1d5b5e4fea","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-05 17:48:03.000000000","message":"Patch Set 3:\n\nFailing on functional tests using master u-c due to, needs new oslo.cache version blacklisting in u-c\n\n  2020-03-05 17:40:56.635 15219 CRITICAL keystone [req-fae444db-f76f-4534-9a9c-38471fbe0604 - - - - -] Unhandled error: TypeError: __init__() got an unexpected keyword argument \u0027dead_retry\u0027\n  2020-03-05 17:40:56.635 15219 ERROR keystone Traceback (most recent call last):\n  2020-03-05 17:40:56.635 15219 ERROR keystone   File \"/openstack/venvs/keystone-testing/lib/python3.6/site-packages/oslo_cache/_memcache_pool.py\", line 150, in _get\n  2020-03-05 17:40:56.635 15219 ERROR keystone     conn \u003d self.queue.pop().connection\n  2020-03-05 17:40:56.635 15219 ERROR keystone IndexError: pop from an empty deque","accounts_in_message":[],"_revision_number":3},{"id":"ac2863ef22e9e2c5992f64a5e3833874cbd4de9d","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-03-05 17:49:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"0c14c86433ed705f3b718e6d989e4a561dd96435","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-05 21:26:50.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4eb2afcc300a423a8c1a3a32252a87ba : SUCCESS in 7m 47s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9b03913291cb4d88b153bac60a858890 : SUCCESS in 1h 24m 51s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c31bc88ccd91450fb7c99e913ef107f9 : TIMED_OUT in 3h 08m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/3f88c54822dd467c978cd8398dba5afd : SUCCESS in 1h 23m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ae2c443d3632487ebcb6335259ef2eff : SUCCESS in 1h 16m 04s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7e42b876e9ec479686a7b431b4203d05 : FAILURE in 1h 14m 54s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0495c919c705429da45f3114ff174cba : FAILURE in 58m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4106b3d645bf43b9b167022a35af16d1 : SUCCESS in 1h 01m 11s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/def32312fef34cb18ec4361053f27f08 : FAILURE in 31m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/47c7631547c447eda11b1af39e0c8a93 : SUCCESS in 54m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90992e129b5d4090a077ff474744eba4 : SUCCESS in 7m 07s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/8af5fc4423f249f99d0a0238adf82346 : FAILURE in 35m 34s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bc917a38d8124f1dab7e428fe8f708d8 : FAILURE in 34m 22s","accounts_in_message":[],"_revision_number":3},{"id":"2809ba1f10fdda7b994d0adc37827a396bd2d901","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-05 21:59:24.000000000","message":"Patch Set 3:\n\nhttps://review.opendev.org/#/c/711427/ must merge before this patch will pass","accounts_in_message":[],"_revision_number":3},{"id":"6aa6da1a1c7c14adbe059f242c1ba685b0e00bbb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-06 06:58:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"e8f00023e9b67181072dd4d7507a03cd6090c74d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-06 08:39:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/229b39446ddd4f1886382f6c25926726 : SUCCESS in 7m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/85c308ee3a964a2488093d627b39ff66 : SUCCESS in 1h 03m 43s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/17e422909fab43dfb938ca531a040125 : SUCCESS in 1h 25m 39s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/35621fec36db446795ffa7d55a85a00f : SUCCESS in 1h 21m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d201e40112294903b8b7279fe7dbcf99 : SUCCESS in 1h 05m 23s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0d370e68ffd94bf2933e713dd62256cc : FAILURE in 1h 21m 57s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3e8d29e41d274d81a06ea46bc0e33920 : FAILURE in 58m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d07e57f8371e4369ad65abae3ef8cc2a : SUCCESS in 1h 36m 38s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/155193ed51654a73a14f8712c22d9eec : FAILURE in 32m 23s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b167643d526546119bbe01452189f078 : SUCCESS in 1h 09m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d9214d5b1764cd69e84ace7ab627471 : SUCCESS in 7m 31s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/6fa83de134f84ee695c943d1d69e6a46 : SUCCESS in 38m 14s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e086364c29704bcda208f2f5a98dd16f : SUCCESS in 35m 33s","accounts_in_message":[],"_revision_number":3},{"id":"30d2081d699fd414df4f605ced4b867bebf571f7","author":{"_account_id":28008,"name":"Guilherme","display_name":"Guilherme Steinmüller","email":"gsteinmuller@vexxhost.com","username":"guilhermesp"},"date":"2020-03-13 22:04:10.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"9a9d3ab66abedf4354bd265d71bb788221a27b92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 22:04:22.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1f486a55503bfd93f6ad7b988e5e70880239e859","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 23:44:17.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3e25ff294baf49a29e789db969300d4c : SUCCESS in 7m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a9abc95b5b2c4a63b2c5fcdd7a9ced97 : SUCCESS in 1h 10m 15s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/51bc784fa64744a3b74c299d241394a2 : SUCCESS in 1h 28m 29s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/85026259080a445ba14c98021c8f0e9c : SUCCESS in 1h 24m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95448dcf2067466da251d14cb006fc0a : SUCCESS in 1h 11m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e5480468ff16411a92910f2355b16628 : SUCCESS in 1h 07m 26s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/513e0be0239f4265ae52f81736df9e62 : SUCCESS in 55m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b57a80ea87c34bfe917365afaf7a7095 : SUCCESS in 5m 02s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/566d332e4ac6471c969aa774604e7f93 : SUCCESS in 43m 18s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4df2a0d1df6846799a9489792eb13982 : SUCCESS in 34m 14s","accounts_in_message":[],"_revision_number":3},{"id":"a70ec5c914892e5c9f3ec626bc5fb804c95c7c1a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 23:44:17.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"515931aecf7fa5b71147f6dbe85f0161bc178122","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-13 23:45:51.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82838c6e201e4753a6ac41031391557e : SUCCESS in 1m 17s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8066ed06543589be52778e01fd64ff72a763748b","revisions":{"edc8011991aeb5947680456fe7784db0d02fd18f":{"kind":"REWORK","_number":1,"created":"2020-03-05 10:47:09.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/711429/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/29/711429/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/29/711429/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/29/711429/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/29/711429/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/29/711429/1"}}},"commit":{"parents":[{"commit":"f4746a362471a3e26df6c2b358db93429acd67f9","subject":"[ussuri][goal] Updates for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f4746a362471a3e26df6c2b358db93429acd67f9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-05 10:33:30.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-05 10:46:59.000000000","tz":0},"subject":"Change from python-memcached to pymemcache","message":"Change from python-memcached to pymemcache\n\nSee https://review.opendev.org/#/c/634457/\n\nChange-Id: Ib34fdb680ecdda6c50be995af2c980aa274f2ba2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/edc8011991aeb5947680456fe7784db0d02fd18f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/edc8011991aeb5947680456fe7784db0d02fd18f"}]},"branch":"refs/heads/master"},"edc060a435e2f5da9880d0b852307f2aba0b2f91":{"kind":"REWORK","_number":2,"created":"2020-03-05 12:41:59.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/711429/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/29/711429/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/29/711429/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/29/711429/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/29/711429/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/29/711429/2"}}},"commit":{"parents":[{"commit":"f4746a362471a3e26df6c2b358db93429acd67f9","subject":"[ussuri][goal] Updates for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f4746a362471a3e26df6c2b358db93429acd67f9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-05 10:33:30.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-05 12:41:55.000000000","tz":0},"subject":"Change from python-memcached to pymemcache","message":"Change from python-memcached to pymemcache\n\nSee [1] for the transition release note. It is necessary to\nkeep both python-memcached and pymemcache as OSA functional tests run\nwith master u-c but integrated tests run from a pinned u-c SHA and the\noslo.cache requirements are currently a moving target [2].\n\n[1] https://review.opendev.org/634457/\n[2] https://review.opendev.org/711427\n\nChange-Id: Ib34fdb680ecdda6c50be995af2c980aa274f2ba2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/edc060a435e2f5da9880d0b852307f2aba0b2f91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/edc060a435e2f5da9880d0b852307f2aba0b2f91"}]},"branch":"refs/heads/master"},"8066ed06543589be52778e01fd64ff72a763748b":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-03-05 12:42:24.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/29/711429/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/29/711429/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/29/711429/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/29/711429/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/29/711429/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/29/711429/3"}}},"commit":{"parents":[{"commit":"f4746a362471a3e26df6c2b358db93429acd67f9","subject":"[ussuri][goal] Updates for python 2.7 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/f4746a362471a3e26df6c2b358db93429acd67f9"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-05 10:33:30.000000000","tz":0},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-03-05 12:42:21.000000000","tz":0},"subject":"Change from python-memcached to pymemcache","message":"Change from python-memcached to pymemcache\n\nSee [1] for the transition release note. It is necessary to\nkeep both python-memcached and pymemcache as OSA functional tests run\nwith master u-c but integrated tests run from a pinned u-c SHA and the\noslo.cache requirements are currently a moving target [2].\n\n[1] https://review.opendev.org/634457\n[2] https://review.opendev.org/711427\n\nChange-Id: Ib34fdb680ecdda6c50be995af2c980aa274f2ba2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8066ed06543589be52778e01fd64ff72a763748b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8066ed06543589be52778e01fd64ff72a763748b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
