)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~714179","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~If2fced0fb579791f249c729d8aaeb5fb50e35620","project":"openstack/openstack-ansible-os_keystone","branch":"master","hashtags":[],"change_id":"If2fced0fb579791f249c729d8aaeb5fb50e35620","subject":"Correct os_user_role parameter from group_name to group","status":"MERGED","created":"2020-03-20 17:17:47.000000000","updated":"2020-03-26 09:29:03.000000000","submitted":"2020-03-23 19:04:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"714179-1584990296389-68d14c44","meta_rev_id":"02f26afb361ae757ccdc4a2dd0d400d0678b5407","_number":714179,"virtual_id_number":714179,"owner":{"_account_id":31750,"name":"Daniel Meloy","email":"danny.meloy@bbc.co.uk","username":"meloyd01"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-03-23 19:04:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-03-23 10:24:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-03-23 10:24:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-03-23 17:11:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-03-23 10:24:37.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-03-23 17:11:35.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-03-23 19:04:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c23bf2c94bcbea35ac9e54fe57b59377abc7471","author":{"_account_id":31750,"name":"Daniel Meloy","email":"danny.meloy@bbc.co.uk","username":"meloyd01"},"date":"2020-03-20 17:17:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1a9501bba953165ae4217133dcdc5b7cc839c532","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-20 19:27:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/08abbcbc195d4928ac5f9617bfec5a9f : SUCCESS in 11m 08s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ddba927e49774c6f900a9ed633cf84dc : SUCCESS in 1h 26m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8bb0cfb07c2240b788c3090df5ce818e : SUCCESS in 1h 38m 23s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/eb288a0a85a543c8b4f8c8864c64e99d : SUCCESS in 1h 34m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8a12ed260c6943e697fe73bc4b974800 : SUCCESS in 1h 17m 27s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2beee43d21234c8a88f1b2c1d8395a11 : FAILURE in 9m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8459445fa44a4561bb7b3ecff46e643a : SUCCESS in 1h 55m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dce45ce8504e4c0e9e4e9771dab92030 : SUCCESS in 1h 05m 42s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0c481fd6e8724e45b6b22899ac3692bf : FAILURE in 37m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/034257b5865247219b3a382a7b916744 : SUCCESS in 59m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a3b5eafa40c47e880e5e44cc792b864 : SUCCESS in 7m 03s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/5108d5a86a8741fca2952add28cae3be : SUCCESS in 41m 19s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4103249ea4a42d3beecb1bf77f0265d : SUCCESS in 51m 15s","accounts_in_message":[],"_revision_number":1},{"id":"4f0a4ccfc2599f14e8f3b67cdc2cbfbc86190e1c","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-23 10:24:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"dba21ab714029c49a6dcb67dd6a8fd31b0323bc5","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-03-23 10:24:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c4818fc814010fac55a2f0c3c0409440531fe47a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-03-23 17:11:35.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"559c835935f89eac6585dc956335bd6b731e7caf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 17:11:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5c416729da506332d0102ad6bb3db489f66a9606","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 19:04:56.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a5a9fb3401b244a78cf76e8b6d30a282 : SUCCESS in 7m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/16af2fdc1cb142d699c4ed4caf05f278 : SUCCESS in 1h 14m 06s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a9d7c77e663a4b88837a19e11538b2f5 : SUCCESS in 1h 49m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/40d9235f1610420082090e1934440c27 : SUCCESS in 1h 18m 38s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0c6a5fdae3dd43a993089c9536bada70 : SUCCESS in 1h 09m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9e25458dc07849ab923faf6e5cb2026f : SUCCESS in 1h 06m 36s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ad0ce01a8c145048c9ad9ebab3a645a : SUCCESS in 58m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35fec26706fa48db8d1468715ef382b2 : SUCCESS in 5m 16s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/db41c413ae704c7aaeed90b21f2e9ec9 : SUCCESS in 38m 59s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f9474c96c49a477a87f9073f857e7ca5 : SUCCESS in 34m 47s","accounts_in_message":[],"_revision_number":1},{"id":"4b7fb1a64b455e7e4037c15da1261f659657d713","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 19:04:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"3918b4c6d11e866200ade8c3bc1e88dc736f4d0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-03-23 19:06:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dba56044af964a6aa0cea639f50b109b : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1},{"id":"02f26afb361ae757ccdc4a2dd0d400d0678b5407","author":{"_account_id":31750,"name":"Daniel Meloy","email":"danny.meloy@bbc.co.uk","username":"meloyd01"},"date":"2020-03-26 09:29:03.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 2086557b66b45179234b990ee92f8d70e71bd070","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff650e4ef8da503b8b94b456f528efbbd914b5ec","revisions":{"ff650e4ef8da503b8b94b456f528efbbd914b5ec":{"kind":"REWORK","_number":1,"created":"2020-03-20 17:17:47.000000000","uploader":{"_account_id":31750,"name":"Daniel Meloy","email":"danny.meloy@bbc.co.uk","username":"meloyd01"},"ref":"refs/changes/79/714179/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/79/714179/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/714179/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/714179/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/714179/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/79/714179/1"}}},"commit":{"parents":[{"commit":"16fe0db40beda6b4a61a63308f508545ae90d38f","subject":"Merge \"Refactor memcached_servers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/16fe0db40beda6b4a61a63308f508545ae90d38f"}]}],"author":{"name":"Danny Meloy","email":"danny.meloy@bbc.co.uk","date":"2020-03-20 17:10:50.000000000","tz":0},"committer":{"name":"Danny Meloy","email":"danny.meloy@bbc.co.uk","date":"2020-03-20 17:10:50.000000000","tz":0},"subject":"Correct os_user_role parameter from group_name to group","message":"Correct os_user_role parameter from group_name to group\n\nChanged the parameter to the os_user_role module from incorrect \u0027group_name\u0027\nto \u0027group\u0027 in line with the ansible docs:\nhttps://docs.ansible.com/ansible/latest/modules/os_user_role_module.html\n\nChange-Id: If2fced0fb579791f249c729d8aaeb5fb50e35620\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ff650e4ef8da503b8b94b456f528efbbd914b5ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ff650e4ef8da503b8b94b456f528efbbd914b5ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
