)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~716971","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Iacaa75b4bb077f25632ec0eecbdf5fd57bb959b2","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"ldap-python-interpreter-fix","hashtags":[],"change_id":"Iacaa75b4bb077f25632ec0eecbdf5fd57bb959b2","subject":"Same service setup host python interpreter for LDAP domains","status":"MERGED","created":"2020-04-02 12:43:19.000000000","updated":"2020-04-02 22:53:40.000000000","submitted":"2020-04-02 22:43:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"716971-1585867414681-e8a3316c","meta_rev_id":"f828205deb430390391228cc9dfa2f8b0f68c6e9","_number":716971,"virtual_id_number":716971,"owner":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-02 22:43:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-04-02 16:56:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2020-04-02 13:07:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-02 20:26:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-02 20:26:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-02 13:07:50.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-04-02 16:56:19.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-04-02 20:26:20.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-04-02 22:43:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2469a9f6a46cad5f8b10632d64fbddb8eb9dae74","author":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"date":"2020-04-02 12:43:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f7c2b8e1b3331a32187cfbcf3f40f2938be81d74","author":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"date":"2020-04-02 12:46:26.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"e669eff873f0d3c03ab85eaefc068e1c6ddf865c","author":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"date":"2020-04-02 12:50:06.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit c03d096ad413040f5b0a3cd3a0397fe974ad37ea","accounts_in_message":[],"_revision_number":2},{"id":"7467cc1b1f84acf5410daa5e488ddf0d9f09b78a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-02 13:07:50.000000000","message":"Patch Set 2: Code-Review+2\n\nLGTM, thanks for patching it!","accounts_in_message":[],"_revision_number":2},{"id":"24c518b40a783084590d687b47541771e3c72269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 15:20:28.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/242c29cd09bf4058a6af2d6f68168235 : SUCCESS in 12m 02s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/95a87c199f63466a9447cdf5a9afef9e : FAILURE in 1h 29m 47s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4c7b4cf0a392426fae7dcf331e46f7ff : SUCCESS in 1h 57m 58s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c998e1ab6d614e5a96cd598011d7ba42 : SUCCESS in 1h 38m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f8f116254e3b40efa7dea8370f067e94 : SUCCESS in 1h 47m 14s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ae27619974994a5daf97854c838cbcfe : FAILURE in 17m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/680bea5f7a3f43a696e852bede246a1c : SUCCESS in 2h 28m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9eed57081c5e40bc830885b003bf3659 : SUCCESS in 1h 46m 52s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7a572397d9f84d9a820b0a31e1809db8 : FAILURE in 57m 04s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8082a72a56a4c4788d5f188816719ac : SUCCESS in 1h 38m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6624447c04af4b9b96728133fe5ab15d : SUCCESS in 7m 53s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/6944cc874c76469a891eceeb77f6a6d2 : SUCCESS in 59m 46s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/166dd438cf06490db90d3d76007d3129 : SUCCESS in 51m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3098d25b3373f914bd20b019b4d53c2ec90c3d68","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-04-02 15:32:34.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6bf5af4832b66f1c4272cfca25665479ba1a23d5","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-04-02 16:56:19.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"dbe4f1fac797e7ddf11436e58ccdc5ff54c09b2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 19:21:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/9dced6186ea54046ac96977c20529a2a : SUCCESS in 7m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/58ed13a5adc740d796c83fd15dfcae74 : SUCCESS in 1h 09m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/073c73deadde4427a7d6814c3b170493 : SUCCESS in 1h 30m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b72d93b2ef344987b134cf3b7cd523b0 : SUCCESS in 1h 25m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5febedae79984408b23272e568712ca6 : SUCCESS in 1h 17m 26s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a046ce2cc21b4e7b94c9b6cb0a363d87 : SUCCESS in 2h 26m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa03979b2a1e47348ed20435887ca377 : SUCCESS in 1h 55m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1a5671464d554f3e8500ec3e3686dc9c : SUCCESS in 1h 07m 51s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b2f8eda29bbf4d87ba27e8a237f51280 : FAILURE in 34m 15s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b420d014afe846a7874b95f48529e47b : SUCCESS in 59m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e040f24f33c24c7cafbaec0a45502231 : SUCCESS in 7m 34s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/eae54773a73f4419aa349e8692449f0f : SUCCESS in 41m 42s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2b04540de8a244818a4636e634b81f59 : SUCCESS in 35m 06s","accounts_in_message":[],"_revision_number":2},{"id":"a39571f01640eb3ac1e013350b6598e1f2cffee6","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-02 20:26:20.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"65c657c12a03762d12f7d9c1dff4dd2e85f1d4f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 20:26:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aa86c663909f67cf223de2865158866b37b9f6f1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:43:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"fcec39a74f26ad70df137cb0d1fed8a4e2d543f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:43:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/18e9b568061c4dd8a0016bd1e7e0e9b5 : SUCCESS in 9m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e562e16b4ba646bfacceff5bb626a8fc : SUCCESS in 1h 11m 10s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9d0d719600c64582ba266d4b58005c95 : SUCCESS in 1h 48m 09s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cd379097c4804727b6af1930b79b0726 : SUCCESS in 1h 31m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69d8661e0a50426ea1357c72610291c3 : SUCCESS in 1h 20m 39s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8d75d09c37af480abc41dbf9120a2d75 : SUCCESS in 1h 19m 45s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/aecfcec05f8a42c7bb365a246847e90b : SUCCESS in 1h 03m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/689329ea1b074349b0422e98c3986bc3 : SUCCESS in 8m 40s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/16fa7ec893524a3389438ec3b90c0d15 : SUCCESS in 41m 07s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/498ed9eb1ae347a4b609580bff557008 : SUCCESS in 37m 41s","accounts_in_message":[],"_revision_number":2},{"id":"1273f0e751320f88193a0678d32e5568d029107a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-02 22:53:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb42ae1d363f4b5d83bb148d89a1d7ea : SUCCESS in 2m 07s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"32b2042d5f22325cb9c6f835cdc872edf3096ccb","revisions":{"3ec8a612ed03487758dd9c028360fc0989a9056f":{"kind":"REWORK","_number":1,"created":"2020-04-02 12:43:19.000000000","uploader":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"ref":"refs/changes/71/716971/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/71/716971/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/716971/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/716971/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/716971/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/716971/1"}}},"commit":{"parents":[{"commit":"ff650e4ef8da503b8b94b456f528efbbd914b5ec","subject":"Correct os_user_role parameter from group_name to group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ff650e4ef8da503b8b94b456f528efbbd914b5ec"}]}],"author":{"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","date":"2020-04-02 12:33:54.000000000","tz":120},"committer":{"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","date":"2020-04-02 12:43:08.000000000","tz":120},"subject":"Use the service setup host python interpreter for LDAP domain creation","message":"Use the service setup host python interpreter for LDAP domain creation\n\nThe service setup host python interpreter has been made configurable and\nthe same needs to be used to setup a Keystone LDAP domain.\n\nChange-Id: Iacaa75b4bb077f25632ec0eecbdf5fd57bb959b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3ec8a612ed03487758dd9c028360fc0989a9056f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3ec8a612ed03487758dd9c028360fc0989a9056f"}]},"branch":"refs/heads/master"},"32b2042d5f22325cb9c6f835cdc872edf3096ccb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-04-02 12:46:26.000000000","uploader":{"_account_id":31792,"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","username":"marcus.klein"},"ref":"refs/changes/71/716971/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/71/716971/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/716971/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/716971/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/716971/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/71/716971/2"}}},"commit":{"parents":[{"commit":"ff650e4ef8da503b8b94b456f528efbbd914b5ec","subject":"Correct os_user_role parameter from group_name to group","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ff650e4ef8da503b8b94b456f528efbbd914b5ec"}]}],"author":{"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","date":"2020-04-02 12:33:54.000000000","tz":120},"committer":{"name":"Marcus Klein","email":"marcus.klein@open-xchange.com","date":"2020-04-02 12:45:37.000000000","tz":120},"subject":"Same service setup host python interpreter for LDAP domains","message":"Same service setup host python interpreter for LDAP domains\n\nThe service setup host python interpreter has been made configurable and\nthe same needs to be used to setup a Keystone LDAP domain.\n\nChange-Id: Iacaa75b4bb077f25632ec0eecbdf5fd57bb959b2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/32b2042d5f22325cb9c6f835cdc872edf3096ccb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/32b2042d5f22325cb9c6f835cdc872edf3096ccb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
