)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~718632","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Ice2434cc0b76024611cf832e8755e05b30ebfe28","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"osa-collections","hashtags":[],"change_id":"Ice2434cc0b76024611cf832e8755e05b30ebfe28","subject":"Use ansible openstack collection","status":"MERGED","created":"2020-04-09 09:29:13.000000000","updated":"2020-06-08 20:56:33.000000000","submitted":"2020-06-08 20:53:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"718632-1591649629114-61bf20c8","meta_rev_id":"4bbeb199d1f63ef8864867ce0a89a554bd0627af","_number":718632,"virtual_id_number":718632,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-08 20:53:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":2,"date":"2020-06-08 19:05:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":2,"date":"2020-06-03 10:29:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"all":[{"value":1,"date":"2020-06-08 19:05:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-03 10:29:08.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-06-08 19:05:09.000000000","updated_by":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"reviewer":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"state":"REVIEWER"},{"updated":"2020-06-08 20:53:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1e8e79126361d50ba2c9e9c7834da4943d45acfa","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-09 09:29:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bea1825849299eb8db7fbfaacbf2a229a46a3aff","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-09 09:42:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d99833451a414b57e1cfe41c9c14f296bf895638","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-09 09:54:44.000000000","message":"Topic set to osa-collections","accounts_in_message":[],"_revision_number":2},{"id":"678c670945fcc2980a47bb6cdbb012a3fd4f3746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 11:31:45.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/67fe6308df964c8297502ca07b20618e : SUCCESS in 10m 25s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/36a5674e8e4948b0bb61beaf1d94f5b0 : SUCCESS in 1h 10m 38s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9c86ae19a08e4b54ba8d6bfbb6f65759 : SUCCESS in 1h 39m 37s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/79439133ec8e497096a2b85f1b6f8294 : SUCCESS in 1h 36m 03s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9d3cef2848dd47458e3e023ff0a827b1 : SUCCESS in 1h 20m 52s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/561a6e4f35a147a2b6d0f888bdd044c1 : FAILURE in 1h 43m 44s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/994cc4b5162647a48b99c8a71c3ec293 : FAILURE in 1h 26m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/1ec308535c934828837ffd0c2e52cd15 : SUCCESS in 1h 31m 22s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/620ea24cba504d3caf433e0d7011bc24 : FAILURE in 39m 38s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8908c69ff81e4cd1884bb0c5798fa2fc : SUCCESS in 56m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4e9bcad187546c2adc1e96b0c0547a6 : SUCCESS in 7m 15s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/35ca19f364c94a3c8efefb6669048bcc : SUCCESS in 38m 11s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/88b7708cdb6849aab8d6956beb9c99a4 : SUCCESS in 35m 44s","accounts_in_message":[],"_revision_number":2},{"id":"74deb192f339973c9a24afc05ad835eb338f1d42","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-04-09 15:04:37.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"94512a4d0dd351d52e0e88c8db6b3eb3c1474f5d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-09 16:50:50.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ad4a53bfbc4d4027a974c62da6bc273d : SUCCESS in 1h 20m 42s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c22b306a9ba241a18664e82739276a4a : SUCCESS in 1h 39m 18s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e6264a8ddeb74ddcb8fe3ca791478c6f : FAILURE in 1h 27m 30s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/06d40f09dd5a47f2b9ab45e898720159 : FAILURE in 41m 15s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/119105fc6e9a436796cbef07b52bcb7f : FAILURE in 1h 36m 33s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af1707261b1c48329db1fd921240d24c : FAILURE in 1h 07m 25s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/19f2306cd9954896965f5450fe3e9a8d : SUCCESS in 1h 10m 39s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c355f525e4974b4ba397dc94015784f2 : FAILURE in 33m 58s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a1b4624f1c194ba194b020fb9421148a : FAILURE in 41m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/340e89aa41df4e8b9c1aa155b77790f7 : SUCCESS in 8m 14s","accounts_in_message":[],"_revision_number":3},{"id":"b49b6a26fb1037d53eb2cff9ecb8c115e67b9afd","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-20 12:39:39.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"7f96d1f05c4e529ab7c1d9b13da7f876bba027f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 15:13:06.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a7539d81f3b64c90950cc45dbb5bdfbd : SUCCESS in 9m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5b588af2c18345adb09b6effa98711da : SUCCESS in 1h 33m 34s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f84407b084274f73b5dc16bb265bf6fb : SUCCESS in 1h 19m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dee4a249d5ab4bccbdd82fac9aa75311 : SUCCESS in 1h 26m 59s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/79c33822c98644c799a00d2ec8e16e5a : SUCCESS in 2h 29m 53s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8d88cd0aba5342388f72da6d50d24f65 : SUCCESS in 1h 58m 31s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2f3d3f6845c6487c9c75da1d539ba2c4 : SUCCESS in 1h 14m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/54691ed39a8e4543b931c26ed853b741 : SUCCESS in 1h 00m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7bb0c86aa187496fbce234aebedc29f9 : SUCCESS in 7m 38s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/22779191c9474b92a907696d8508a04b : SUCCESS in 41m 19s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/dc1dc6cd92914ff68b123bbb95523f36 : SUCCESS in 36m 00s","accounts_in_message":[],"_revision_number":4},{"id":"a09e1653bbc44a9b9c6a7b8178eaf92c87b2818a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-06-03 10:29:08.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"ca41456a731767d643748f47d917022945beb946","author":{"_account_id":13095,"name":"Marc Gariépy","email":"gariepy.marc@gmail.com","username":"mgariepy"},"date":"2020-06-08 19:05:09.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"706bfc856e628a23c551220906e4cf0e29f6d780","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 19:05:21.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"01ef60fa228923e718f5c790bf8f6301193cb838","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:53:48.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dfcaf268e75742188892c906b67d5fd0 : SUCCESS in 10m 09s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/bf428fca9d004dc9bb63ece89161fd30 : SUCCESS in 1h 30m 22s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/60673b4a2c10461db9955b52bd7892e2 : SUCCESS in 1h 44m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56aa5fa8ba42467db664f0e832043d3f : SUCCESS in 1h 27m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dea2165ff71646c2bbbd265bd1e315b2 : SUCCESS in 1h 08m 44s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e89da01ac4f417f8244dafc57168dde : SUCCESS in 1h 03m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f2ffbd79bb045c6afa83e2455c32f10 : SUCCESS in 7m 25s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/0c7b754e56fc45e895cedc5d9c1eda98 : SUCCESS in 41m 39s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0128182616d949b88b1cba8d182c2ea4 : SUCCESS in 41m 15s","accounts_in_message":[],"_revision_number":4},{"id":"dc764d321e42ebd956cf7cf0dd6bad2e9b30e9a3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:53:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"940c723b107ede770913df8fe41b7110c73c0cb9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-08 20:55:22.000000000","message":"Patch Set 4: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 39364dd53c6387308cbf826cfde30c066988fb35","accounts_in_message":[],"_revision_number":4},{"id":"4bbeb199d1f63ef8864867ce0a89a554bd0627af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:56:33.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e729572a53ff4fa0a3c7d30e9f5c4149 : SUCCESS in 2m 19s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8141fb5ef9890e20cbbc48150ea83ca3f423e2e1","revisions":{"a6727a7eebdbb033d00ced2ffd0979263a12d765":{"kind":"REWORK","_number":1,"created":"2020-04-09 09:29:13.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/32/718632/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/32/718632/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/1"}}},"commit":{"parents":[{"commit":"32b2042d5f22325cb9c6f835cdc872edf3096ccb","subject":"Same service setup host python interpreter for LDAP domains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/32b2042d5f22325cb9c6f835cdc872edf3096ccb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:27:44.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:29:09.000000000","tz":60},"subject":"Use ansible openstack collection.","message":"Use ansible openstack collection.\n\n3 uses of the OSA \u0027keystone\u0027 module remain and should be replaced\nin a future patch when the collection contains a suitable module.\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: Ice2434cc0b76024611cf832e8755e05b30ebfe28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a6727a7eebdbb033d00ced2ffd0979263a12d765"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a6727a7eebdbb033d00ced2ffd0979263a12d765"}]},"branch":"refs/heads/master"},"d11e1bf7a23d03fba27d87973ab3581894b46b95":{"kind":"REWORK","_number":2,"created":"2020-04-09 09:42:37.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/32/718632/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/32/718632/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/2"}}},"commit":{"parents":[{"commit":"32b2042d5f22325cb9c6f835cdc872edf3096ccb","subject":"Same service setup host python interpreter for LDAP domains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/32b2042d5f22325cb9c6f835cdc872edf3096ccb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:27:44.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:42:06.000000000","tz":60},"subject":"Use ansible openstack collection.","message":"Use ansible openstack collection.\n\n3 uses of the OSA \u0027keystone\u0027 module remain and should be replaced\nin a future patch when the collection contains a suitable module.\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: Ice2434cc0b76024611cf832e8755e05b30ebfe28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d11e1bf7a23d03fba27d87973ab3581894b46b95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d11e1bf7a23d03fba27d87973ab3581894b46b95"}]},"branch":"refs/heads/master"},"4d61d7d7861414b56c77a3ee9866c49e5d3ac75c":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-04-09 15:04:37.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/32/718632/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/32/718632/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/3"}}},"commit":{"parents":[{"commit":"32b2042d5f22325cb9c6f835cdc872edf3096ccb","subject":"Same service setup host python interpreter for LDAP domains","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/32b2042d5f22325cb9c6f835cdc872edf3096ccb"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:27:44.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 15:04:35.000000000","tz":0},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\n3 uses of the OSA \u0027keystone\u0027 module remain and should be replaced\nin a future patch when the collection contains a suitable module.\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: Ice2434cc0b76024611cf832e8755e05b30ebfe28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/4d61d7d7861414b56c77a3ee9866c49e5d3ac75c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/4d61d7d7861414b56c77a3ee9866c49e5d3ac75c"}]},"branch":"refs/heads/master"},"8141fb5ef9890e20cbbc48150ea83ca3f423e2e1":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-05-20 12:39:39.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/32/718632/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/32/718632/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/32/718632/4"}}},"commit":{"parents":[{"commit":"4d63f8b9d870f61bcf2130114effe0ca3369bde5","subject":"Merge \"[ussuri][goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/4d63f8b9d870f61bcf2130114effe0ca3369bde5"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-04-09 09:27:44.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-05-20 12:39:39.000000000","tz":0},"subject":"Use ansible openstack collection","message":"Use ansible openstack collection\n\n3 uses of the OSA \u0027keystone\u0027 module remain and should be replaced\nin a future patch when the collection contains a suitable module.\n\nDepends-On: https://review.opendev.org/718431\nDepends-On: https://review.opendev.org/718362\nChange-Id: Ice2434cc0b76024611cf832e8755e05b30ebfe28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8141fb5ef9890e20cbbc48150ea83ca3f423e2e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8141fb5ef9890e20cbbc48150ea83ca3f423e2e1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
