)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~727758","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Ftrain~Ia978b46f6a6dfc5da8f8ebecb1a3c9fe44948add","project":"openstack/openstack-ansible-os_keystone","branch":"stable/train","topic":"mod-auth-openidc-cache-stable/train","hashtags":[],"change_id":"Ia978b46f6a6dfc5da8f8ebecb1a3c9fe44948add","subject":"Add memcached caching","status":"MERGED","created":"2020-05-13 13:21:48.000000000","updated":"2020-05-17 10:47:27.000000000","submitted":"2020-05-17 10:46:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727758-1589712374089-7d7648ba","meta_rev_id":"23e459324bfdf5993e88540047fd4c99c0c2c5ab","_number":727758,"virtual_id_number":727758,"owner":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-17 10:46:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"recommended":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"all":[{"value":1,"date":"2020-05-13 13:52:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":2,"date":"2020-05-13 13:45:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-15 21:03:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-15 21:03:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 13:45:29.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-05-13 13:52:46.000000000","updated_by":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"reviewer":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"state":"REVIEWER"},{"updated":"2020-05-15 21:03:36.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-05-17 10:46:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"340e4fab169db0e1fea583431c191667d37e4736","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-05-13 13:21:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9bdaca729d409748578dc37faef54b90a875fc45","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-13 13:45:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6fa3fd7ec4bf1ed7fc75e75d73e469ac4297ab22","author":{"_account_id":27336,"name":"XiaojueGuan","email":"xiaojueguan@gmail.com","username":"XiaojueGuan"},"date":"2020-05-13 13:52:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9ebb48f805d17a209d8fa7741ebf87cff5126503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 15:55:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dd72901f3de646078e605722a60e16fb : SUCCESS in 6m 46s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b2a3494d9113474592115f1d28d245a1 : SUCCESS in 1h 09m 09s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/78066fa0cc6a40a8839d8486fede0169 : SUCCESS in 1h 26m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/ea3fce1f7f5743f6a71de5c59d4a5be4 : SUCCESS in 1h 32m 33s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/c0e9c7435fe3452aaf8861b1c33dc824 : SUCCESS in 1h 21m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ca2c211c5e9465a9311989e67ef24ea : SUCCESS in 1h 17m 57s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d26b7559345e4283a101a5f8f829d3df : SUCCESS in 2h 18m 45s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a868f7dc8ca0412b919180df50536712 : SUCCESS in 1h 44m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a19f219a923c485b8d77f6b430633fc5 : SUCCESS in 1h 15m 12s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/47cc85f2b011405a9bb648de3f6af5c5 : FAILURE in 32m 26s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7ba4fd82902c47009ccc04150eabb5a1 : SUCCESS in 58m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b607c063191a41a2807e3fd1311fb1b5 : SUCCESS in 5m 42s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/df566696ceb54acba72ed00107856c4b : SUCCESS in 36m 04s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5bb57f84e3ba4307a7772553a753a7d1 : SUCCESS in 41m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c014b1a499f88ab385ef6d6c4a914ea06dc275b7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-05-15 21:03:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"c607b9f871ffeccf5e27a48576689ce13aea631d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 21:03:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e10f7adf4dabe3727369f824806ceb964aa887ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-15 22:54:56.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2edbdcb0aff648098f94f1ac4665219f : SUCCESS in 10m 53s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/2cd5ef900f96491fb29703320c2503ea : SUCCESS in 1h 04m 12s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/980b521acf5d4488a5367f72d0362e7e : SUCCESS in 1h 23m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/834ce58371294e2694f0f06c65ca1f1b : SUCCESS in 1h 36m 19s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f308733fde5a4982b5d6a0056cc2d6dc : FAILURE in 1h 30m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c12577e29dbb4afe9694b6e080ceb966 : SUCCESS in 1h 11m 38s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c98728494f444798bfd0d869e77f2bbe : SUCCESS in 1h 42m 05s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5eb77d318721450c8861f6675d22136e : SUCCESS in 1h 15m 24s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4c14b1383f1347a7a4cf0a1be20ac5f7 : SUCCESS in 58m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b57153d9e6a4041b47f2a6705abfbd6 : SUCCESS in 9m 23s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/b9df5ba57aae48a18022b32e5407e7e7 : SUCCESS in 37m 25s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62d06af09d944df1aa26aedf8470f413 : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0a627248e34024e92dd26dae986cd45d7937f358","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-05-17 06:48:26.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f57de98d014e714d9b9e46c0d9b4d250a7e07624","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 08:57:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6ab0b71e712f42deb18451db65745d71 : SUCCESS in 5m 05s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f029c06bc5ad464b9430e7fd68027428 : SUCCESS in 55m 50s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/abc3004f7a3348a39f4552b45e4571a2 : SUCCESS in 1h 14m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a0b6f7e8c3b14da9af757011af433a1c : SUCCESS in 1h 22m 56s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f013958579be49efa5502561c582ccc3 : SUCCESS in 1h 24m 54s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f7795b7eeb324bffb51713141c656ad6 : SUCCESS in 1h 16m 48s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/48dabe9c598c441b906897bcd87ea28d : SUCCESS in 2h 07m 03s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9ff4218f4d70487887180adf1d0ed70f : SUCCESS in 1h 38m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e0bd946500684f969a448858c7bb37d9 : SUCCESS in 56m 48s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/064a9c9fb2844220a9a396eca2362fff : FAILURE in 28m 53s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a00706f8e9c049c484ff275bb6f5df28 : SUCCESS in 55m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0499693af4214693a552a7980c9952c6 : SUCCESS in 5m 23s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/2371cbd648fd40c89d3688cc98ae5fdc : SUCCESS in 35m 19s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0ac6b7007d1344f18240027f702855d7 : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":1},{"id":"97d2f8d74cc666044723060a9aa540f0c7e50ff5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 08:58:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ec99883627527f0a3f4b5a7f99f86811ae7f59d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 10:46:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a7c12ee92dde4beabb23063a9955be36 : SUCCESS in 6m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/89a67b9201c14c8298cc4be940ee44f7 : SUCCESS in 1h 04m 39s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/9f97bc89ea09407ca652c628edc50962 : SUCCESS in 1h 18m 16s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a5a3110e9198409ab1b022d525147315 : SUCCESS in 1h 34m 30s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/8a9fe313c60442bcad6d6d628ec113c4 : SUCCESS in 1h 27m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/769c2e1b286c4f75a8e7cd16c721ae74 : SUCCESS in 1h 19m 30s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d2e8abf8ba7c4c51a54c5a361f73e86f : SUCCESS in 1h 46m 08s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/48fc6a3b56da4b8ca8b5d679073836b4 : SUCCESS in 1h 13m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d79bec7152c4ade84114ab61a2ec369 : SUCCESS in 58m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83a54088f4514466b80fb0f9fccff342 : SUCCESS in 6m 14s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/afbd8d724dad4f6087c3bc1abddc6e31 : SUCCESS in 35m 59s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/48b231b47089456f9ee96de82d551666 : SUCCESS in 32m 07s","accounts_in_message":[],"_revision_number":1},{"id":"639417b4ba14f0c16853bd0e2a23836c9994ea94","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 10:46:14.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2b1467d03732d2de3726575ecbc1142ca911b6bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-17 10:47:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d95dec9581a14de98bcf8dba65f24ffe : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"579fd7bed69548884bf92aed3ba60eaab9cdb92a","revisions":{"579fd7bed69548884bf92aed3ba60eaab9cdb92a":{"kind":"REWORK","_number":1,"created":"2020-05-13 13:21:48.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/58/727758/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/58/727758/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/58/727758/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/58/727758/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/58/727758/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/58/727758/1"}}},"commit":{"parents":[{"commit":"43fa704005b13e6665e3e4a41b012407453633dd","subject":"Merge \"Add option for OIDCOutgoingProxy for mod_auth_openidc\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/43fa704005b13e6665e3e4a41b012407453633dd"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-04-28 17:11:01.000000000","tz":60},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-05-13 13:21:48.000000000","tz":0},"subject":"Add memcached caching","message":"Add memcached caching\n\nPiggybacking on to the memcached containers that keystone uses for\nits token caching we are able to share state between the apache_mod_openidc\ninstances so that authentications succeed in cases where auth requests\nare being routed to more than one keystone container.\n\nChange-Id: Ia978b46f6a6dfc5da8f8ebecb1a3c9fe44948add\n(cherry picked from commit beebf1196e38ce76e0a65ff46c2bb58585d8e5bd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/579fd7bed69548884bf92aed3ba60eaab9cdb92a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/579fd7bed69548884bf92aed3ba60eaab9cdb92a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
