)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~735869","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Ftrain~I020986bbc2d5baa73a19ee7e1070019cb4e9ce63","project":"openstack/openstack-ansible-os_keystone","branch":"stable/train","topic":"add-oidc-auth-params-stable/train","hashtags":[],"change_id":"I020986bbc2d5baa73a19ee7e1070019cb4e9ce63","subject":"Add OIDCAuthRequestParams parameter to template","status":"MERGED","created":"2020-06-16 10:56:25.000000000","updated":"2020-07-06 18:12:10.000000000","submitted":"2020-07-06 18:08:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735869-1594058938821-abf989b5","meta_rev_id":"90895d99be73319c4beb5814da8919c08f4b7b29","_number":735869,"virtual_id_number":735869,"owner":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31750,"name":"Daniel Meloy","email":"danny.meloy@bbc.co.uk","username":"meloyd01"},{"value":0,"date":"2020-06-25 13:56:03.000000000","_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-06 18:08:58.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-26 17:10:56.000000000","_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31750,"name":"Daniel Meloy","email":"danny.meloy@bbc.co.uk","username":"meloyd01"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":2,"date":"2020-07-06 13:05:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"date":"2020-06-16 10:56:25.000000000","_account_id":31750,"name":"Daniel Meloy","email":"danny.meloy@bbc.co.uk","username":"meloyd01"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":1,"date":"2020-07-06 15:56:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31750,"name":"Daniel Meloy","email":"danny.meloy@bbc.co.uk","username":"meloyd01"},{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"_account_id":31750,"name":"Daniel Meloy","email":"danny.meloy@bbc.co.uk","username":"meloyd01"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 10:56:25.000000000","updated_by":{"_account_id":31750,"name":"Daniel Meloy","email":"danny.meloy@bbc.co.uk","username":"meloyd01"},"reviewer":{"_account_id":31750,"name":"Daniel Meloy","email":"danny.meloy@bbc.co.uk","username":"meloyd01"},"state":"REVIEWER"},{"updated":"2020-06-26 17:10:56.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-07-06 15:56:22.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-07-06 18:08:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7867fb800f7bae6f321341f8f669f03f3d6820f3","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-16 10:56:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e0cd25864d9f8a9b5147bc16b274dc9bd5e4b79a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 13:03:26.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0fbe4e43e4ae4bd393e58a8bc3b6d2ce : SUCCESS in 11m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/5faab04787fb4cbfbfefef7fce106b57 : RETRY_LIMIT in 15m 57s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/27bc8a21b7ac4b5cbc43689c6399e1d6 : RETRY_LIMIT in 18m 26s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7884bd0852dd469e895f8a93d871da52 : RETRY_LIMIT in 13m 01s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/03c35bb722b24962b964030879de4d60 : RETRY_LIMIT in 13m 49s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbccc7208a9c47a1ad2d0b716e2ef703 : RETRY_LIMIT in 17m 59s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c1a55b8badb143818983cb90fe5eb90d : RETRY_LIMIT in 14m 07s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7096fbdd933c45149b417641d357a044 : SUCCESS in 1h 45m 24s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/02472c518e204e088aad94459033ce65 : RETRY_LIMIT in 12m 50s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/572fe292af6d4ba8b6819c3854826136 : RETRY_LIMIT in 10m 52s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3c85e0d8592e40f8a75eb91f7ddc3950 : RETRY_LIMIT in 15m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c5a073c62b54ab88c9723b91dce1d0a : SUCCESS in 9m 25s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/79ec0d56faa24c0bbb229ed17ceabfca : FAILURE in 21m 41s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/186c4d9f665e46d79918cffa58d5fd71 : FAILURE in 10m 26s","accounts_in_message":[],"_revision_number":1},{"id":"3f3a29b37d8fa373e5d329fdaa7fc327399efc71","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-17 19:43:01.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dbe70e4430e713132433059ed3ad7101d7f4a11b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 21:24:11.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6b6a3147641b4df3ae4e027744514761 : SUCCESS in 10m 32s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b875df5f084e4b84ad60bb959e0fedfb : RETRY_LIMIT in 9m 24s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/e1f1c8464c704599b7bc7dbb38269efc : RETRY_LIMIT in 8m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cdaebdc9fd1c46658cc5b5ae7ecd99e2 : RETRY_LIMIT in 12m 24s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/52b716ce21b249eca5dafa247883871e : RETRY_LIMIT in 25m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71a84295bbf54d6da862419f379cce39 : RETRY_LIMIT in 11m 33s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f76f3827b70c4e169255965d783fee61 : RETRY_LIMIT in 13m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f26881f22b8423c9873aa9c69bf5746 : RETRY_LIMIT in 11m 47s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/da6323801eae47b785bb0535e698f932 : RETRY_LIMIT in 10m 21s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fecf66c87bcb4f74b2141d2107ff0d3b : RETRY_LIMIT in 20m 12s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53c4458b9e7e49f2abb6c75ebca9b65a : RETRY_LIMIT in 15m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35f45e6d902b46ca80890e734dda76ca : SUCCESS in 12m 43s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/3f8ecd5635b14cc59ecefae9b2fb9d79 : FAILURE in 15m 28s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/722cd601b48a4a9ca8bd02fcda25fd6f : FAILURE in 10m 33s","accounts_in_message":[],"_revision_number":1},{"id":"9e912d1660f1cb96ed0e7ea06d5587c8be67e277","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-24 10:52:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0d59c4acf513d91860ea5d9c67be817d4a029009","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-24 12:25:57.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dbaf0f09fa384ec4a410149f6a872866 : SUCCESS in 7m 27s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/047f20061fad48ff8de5d4729e8afb72 : RETRY_LIMIT in 6m 20s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/bc69a0c3f0b042d6848925863f509613 : RETRY_LIMIT in 7m 49s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e5e5304cb46947869bbb044cc1816910 : RETRY_LIMIT in 8m 36s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/7ca65fee56f3495fad9aa069f8792777 : FAILURE in 1h 25m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c8ce343c47ce485da495a37358395098 : RETRY_LIMIT in 5m 39s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5e001e0a4de247ceb861a80d36909655 : RETRY_LIMIT in 9m 04s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f905a115cc154eb6be1e8c5e0f06c3a5 : RETRY_LIMIT in 8m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/662c06afca2845628ea51003348b6880 : RETRY_LIMIT in 9m 05s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/d909f67135234af3a824be6033cd5b2e : FAILURE in 34m 34s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/997310bd60d6443dbecffba1091a51ba : RETRY_LIMIT in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/260765b1673e49de9a0ea28fd6041394 : SUCCESS in 6m 34s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/701376025d4a40c1ae154556615b4461 : SUCCESS in 39m 44s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e5cdf2ebe3df4edca73cc8cb8bea060d : SUCCESS in 37m 33s","accounts_in_message":[],"_revision_number":1},{"id":"cf38e391ba955cd668fd54f552b54a2e28f94c87","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-25 13:56:03.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"67ec3a073c92f9288ccbb806d50ba2e22d332113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-25 15:24:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/865ac6e3da0e4d339d47fccd1af6185c : SUCCESS in 7m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/910e4414d11449588f20b1b6bda807e2 : RETRY_LIMIT in 6m 57s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/b3019e7804f243edbe28eaf6a6f30081 : RETRY_LIMIT in 9m 12s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cd47594f79574cc1a72e959f71734fd2 : RETRY_LIMIT in 9m 40s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e77079d9398e4929b34282e13ccf03f2 : FAILURE in 1h 22m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1843e4fe11b24a03946661c6d7123009 : RETRY_LIMIT in 7m 49s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3470548d8e81408890c902ad4ad9a11e : RETRY_LIMIT in 16m 50s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53e44559c43d4a0dbf2cb2c7fb1193ac : RETRY_LIMIT in 11m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a06d33daae7448768786d3fcfc268192 : RETRY_LIMIT in 10m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/768f535c16fe4f00970297b0235ebe3b : FAILURE in 37m 00s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ecdbb30dfe6449109ebe9d8c3478f0b1 : RETRY_LIMIT in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fdf033d422c4cf0b0c3b3c1117dd5f0 : SUCCESS in 6m 30s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/5033cab1a0c64e4e8953c71cd2972a8c : SUCCESS in 36m 23s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d695565bf0054e34aa5e89e2b79c1654 : SUCCESS in 35m 07s","accounts_in_message":[],"_revision_number":1},{"id":"0efc1547c826c1974224675c790ade62aa8242ea","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-26 17:10:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"62fff163148ad5046bac4948c9ab1aa7ed2e1aa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 20:18:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/174a92149c664881a27a1a8e0f19581f : TIMED_OUT in 18m 04s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/3e15937ff61f40af9013ddb40c6ee185 : SUCCESS in 1h 10m 15s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/c9a390dc32e94b518e75c2212170b207 : SUCCESS in 1h 20m 57s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/142ef1c0a0164be289065e21cc176f8a : SUCCESS in 1h 29m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/b6817685caaf4f83b7da51b34a2374bb : FAILURE in 1h 23m 51s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b93e443d4eda45f4b820a969933f8359 : SUCCESS in 1h 33m 43s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f72fb7e7926240098d3f1a82774d6984 : SUCCESS in 2h 54m 10s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02bda9c8acaa44eb8599efb88fe6ccb0 : SUCCESS in 1h 50m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/019df8852bce42de82ac3c5911e5b381 : SUCCESS in 1h 04m 12s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/0b2a72e5bed1497f93ee6643f724b3ab : FAILURE in 38m 21s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e16e5acd62bc4c65887618ef0d171cfb : SUCCESS in 1h 05m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/075eeefe4670495da4878983c733d594 : SUCCESS in 11m 05s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/f031d7294dcd4ac39393ddbc937abf6b : SUCCESS in 37m 13s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fd2d99d36d1e4401aa257594bace793b : SUCCESS in 33m 48s","accounts_in_message":[],"_revision_number":1},{"id":"44cbd09ebc071f11dae279e16929fa29ecd22279","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-06 13:05:48.000000000","message":"Patch Set 1: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"12df208d80b28ba9f3f3561b434984001e075122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:25:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0c3f7b782607453c95e0894c57127dc7 : SUCCESS in 6m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/df1abfc4261b49df88f2b54284270aab : SUCCESS in 1h 04m 01s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/3ff2e8bf026f407d9f0f89be83f7d1d1 : SUCCESS in 1h 17m 02s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d2d081c08d6f4642b698f9ef025907ee : SUCCESS in 1h 29m 04s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/fd79cd99be994f0ba44e77b927c4bb4b : RETRY_LIMIT in 19m 33s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/37d0d42b0b1c4973811f9d74d68995c3 : SUCCESS in 1h 12m 02s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/11633854f7cf476a9190a245d68ba71e : SUCCESS in 2h 14m 35s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0201973c97c44164b1bbb0e51a26b51c : SUCCESS in 2h 02m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/76a77a4684334f45bcc8facf39c37405 : SUCCESS in 1h 13m 39s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cb16ecd7338a44e291de8492d68e3d0c : RETRY_LIMIT in 20m 18s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c31a342bc26d4d4a99316236c948a6af : SUCCESS in 56m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0906fa970f6483485b9d536270b252a : SUCCESS in 15m 51s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/0f59d32dbbc448a4ad85273ca529039d : SUCCESS in 51m 39s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a953b0b4daef4208a546544acc3455af : SUCCESS in 41m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9f369df3c2a4d5da3dd30df7d9f5340af92ccc63","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-06 15:56:22.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6d6678e6a121b2a983047e882cdc8120d34181eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 15:56:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"73774f7c1cee38512dcdbdee6bb9f778f4b0525f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 18:08:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"82da25a8d42966f475b34b9c16f92bb0102c325f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 18:08:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/93a2845597f8431997019df278fecbe8 : SUCCESS in 10m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/647d294a900642cda500b340783ebb3a : SUCCESS in 1h 03m 35s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/7dba96fbb9e24a34a718979a39efc4d7 : SUCCESS in 1h 17m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/829ca4be3d844a76a983c1be0ae63760 : SUCCESS in 1h 24m 55s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cae478a38a364f59b650e7d525af2df2 : RETRY_LIMIT in 19m 30s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/082c520c9377487e8a0909ebfd3ce9a4 : SUCCESS in 1h 12m 46s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19209ac1003242749dd8b4104e743448 : SUCCESS in 2h 00m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a98c1386e2b848269858b56166d349a9 : SUCCESS in 1h 25m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ac3a70fc983447cf97c90f6b840ebdfd : SUCCESS in 57m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2419a373b09f4f4aa7e38df3cfddd596 : SUCCESS in 9m 24s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/9438765adf314ae99c291154e4967e25 : SUCCESS in 38m 17s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/932bea4f800c4e30bcacd1990d17a1dd : SUCCESS in 32m 09s","accounts_in_message":[],"_revision_number":1},{"id":"9fb3f8dcd2da16e228896d20468fd947bc37024e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 18:12:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2925cbd6203346ba89bf419c400fed2b : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c773cc692d00c485f8be1a4dfc62adc905004abf","revisions":{"c773cc692d00c485f8be1a4dfc62adc905004abf":{"kind":"REWORK","_number":1,"created":"2020-06-16 10:56:25.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/69/735869/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/69/735869/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/69/735869/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/69/735869/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/69/735869/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/69/735869/1"}}},"commit":{"parents":[{"commit":"579fd7bed69548884bf92aed3ba60eaab9cdb92a","subject":"Add memcached caching","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/579fd7bed69548884bf92aed3ba60eaab9cdb92a"}]}],"author":{"name":"Danny Meloy","email":"danny.meloy@bbc.co.uk","date":"2020-06-02 12:39:33.000000000","tz":60},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-06-16 10:56:25.000000000","tz":0},"subject":"Add OIDCAuthRequestParams parameter to template","message":"Add OIDCAuthRequestParams parameter to template\n\nAdded the OIDCAuthRequestParams line to the keystone-httpd.conf template\nThis allows for the addition of optional extra parameters that will be sent\nalong with the Authorization Request when using federated logins:\nhttps://github.com/zmartzone/mod_auth_openidc/blob/master/auth_openidc.conf\n\nChange-Id: I020986bbc2d5baa73a19ee7e1070019cb4e9ce63\n(cherry picked from commit f0ce41ea61ed13933157e6d4794cf83040f3fd7a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c773cc692d00c485f8be1a4dfc62adc905004abf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c773cc692d00c485f8be1a4dfc62adc905004abf"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
