)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~738202","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Ftrain~I79a32f696f592b6fb3fd2c01dc9f077597b1a7a9","project":"openstack/openstack-ansible-os_keystone","branch":"stable/train","topic":"idp-improvements-stable/train","hashtags":[],"change_id":"I79a32f696f592b6fb3fd2c01dc9f077597b1a7a9","subject":"Identity providers can be created with specified domain","status":"ABANDONED","created":"2020-06-26 13:14:52.000000000","updated":"2020-07-06 13:12:05.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f48fd9bffc56eafa43f7fd3ae2f8fd7def3bdcb2","_number":738202,"virtual_id_number":738202,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-06-29 15:06:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-26 13:14:52.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-06-29 15:06:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5dd1c158fe47f45e00f2c1a9c03341d46742fe86","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-26 13:14:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"68cd74a67842403db892c06ff50b6a5101068b21","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-26 13:16:55.000000000","message":"Patch Set 1: Workflow-1\n\nTwo commit ID - not good","accounts_in_message":[],"_revision_number":1},{"id":"dc5c6f989ced0e9fe3c5cd3322b1b50e3ca30fd5","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-26 13:17:31.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"53581f64f11b5b3d5862e42b89059509a8c2f3ff","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-06-26 13:24:53.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"3f3747260c4e4a1562c0b60b27316b9581edea44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 14:51:14.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/54b08adcf1614afc93952a2b92ad99e8 : TIMED_OUT in 16m 14s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/f075baa120604199b1161fd71604ae9c : RETRY_LIMIT in 12m 39s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/f4de43dc9e9241648228056b548e33b5 : RETRY_LIMIT in 8m 20s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b5c64f8e5f454d4583ccb992c1f07cae : RETRY_LIMIT in 9m 58s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/2c6a51e2133a4e8686d06ae99b00359f : FAILURE in 1h 23m 56s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/af293f348a4b4289876f919faeb7756a : RETRY_LIMIT in 9m 12s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d0a285d841124e6993d42bc768338ed4 : RETRY_LIMIT in 9m 15s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/993b6f7ffe3a430b9c15b11beb4e1675 : RETRY_LIMIT in 10m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bdbecdb1699e4ac895ac64a3e642ae3f : RETRY_LIMIT in 7m 45s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/e6fcfc0052b7426c928f2c805ceede80 : FAILURE in 36m 43s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3bd568fa4c944312b85344295440608a : RETRY_LIMIT in 7m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ee4774f075445bd9516f95bcbe1509e : SUCCESS in 6m 16s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/41e8a27231c340ed8e8cb04e957bca2c : SUCCESS in 37m 56s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/74b3c0446512414785c361b6ac765794 : SUCCESS in 32m 38s","accounts_in_message":[],"_revision_number":2},{"id":"49166ecc849e6f7a5ad6dbce25610fb5e1467cc2","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-06-29 11:39:45.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"85c70ba31fb2a110ac0b3aea2f2ccdce03e9ec44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-29 15:06:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/5901cc8763f74e6bb0c206eaf546f56e : SUCCESS in 9m 01s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/8eb166afeb2d4a50a868245e6cb1d66e : FAILURE in 1h 17m 56s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/1b1ba3a0d2b4448c816d455b4b4ff7a5 : SUCCESS in 1h 36m 00s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dd3fe017f1de4efe869f8a3f54840853 : SUCCESS in 1h 42m 07s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/80d93e6a16934b9da8ea4d2b08d03220 : FAILURE in 1h 34m 53s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1e2bbb6ccf3c41aea9d494ab7c68dc20 : SUCCESS in 1h 28m 39s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/337729d2ead74d7d9d17216454b726ca : SUCCESS in 2h 33m 59s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a4f881b9d1a84d63adda7d9949e3fbda : SUCCESS in 1h 51m 45s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/a274b5bbd9664ceab526392e3f4cf810 : SUCCESS in 1h 57m 19s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/cbcb371077b04eccb6185154f93cd3b6 : FAILURE in 32m 13s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/91374819e7754c5791122bb2f15065f4 : SUCCESS in 51m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e203043b10104d45af5465b02613246f : SUCCESS in 6m 00s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/56f2ae5d108b44498c76dc4cb7c57da5 : SUCCESS in 35m 48s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0d3823cf08054e97a4fd6a54ef29cbe5 : SUCCESS in 31m 11s","accounts_in_message":[],"_revision_number":3},{"id":"3c8cdf3f382603731f556d33a61506036eddcced","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-06 13:12:05.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"97d17b821e9b60d8b72e4d32af1b2ec58aff149d","revisions":{"fc62c074904de245aeeedb8040908ecbfda5e8f4":{"kind":"REWORK","_number":1,"created":"2020-06-26 13:14:52.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/02/738202/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/02/738202/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/738202/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/738202/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/738202/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/738202/1"}}},"commit":{"parents":[{"commit":"579fd7bed69548884bf92aed3ba60eaab9cdb92a","subject":"Add memcached caching","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/579fd7bed69548884bf92aed3ba60eaab9cdb92a"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-06-15 18:52:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-26 13:14:52.000000000","tz":0},"subject":"Identity providers can be created with specified domain","message":"Identity providers can be created with specified domain\n\nIDPs are linked to a keystone domain. Currently when we provision a new IDP\nit gets created within a new autogenerated domain. Provisioners can now\ngive a domain_id in an IDP entry within the trusted_idp_list and the IDP\nwill be created within that domain.\n\nDepends-On: https://review.opendev.org/734672\nDepends-On: https://review.opendev.org/735289\nDepends-On: If761dd4faf63cf09d436d9e0a22f7e90cec9a105\nChange-Id: I30bcbce4c75b38f96a27ee742d15cf2abe402176\n\nAdd IDP display_name to defaults\n\nAllows operator a name different to the one used in the backend\nto display to the user in the horizon Identity Provider dropdown.\n\nDepends-On: https://review.opendev.org/#/c/736739/\nChange-Id: I79a32f696f592b6fb3fd2c01dc9f077597b1a7a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/fc62c074904de245aeeedb8040908ecbfda5e8f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/fc62c074904de245aeeedb8040908ecbfda5e8f4"}]},"branch":"refs/heads/stable/train"},"ca7002b79b7041e85b57a92e8e53280a08bebc03":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-26 13:17:31.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/02/738202/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/02/738202/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/738202/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/738202/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/738202/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/738202/2"}}},"commit":{"parents":[{"commit":"579fd7bed69548884bf92aed3ba60eaab9cdb92a","subject":"Add memcached caching","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/579fd7bed69548884bf92aed3ba60eaab9cdb92a"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-06-15 18:52:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-06-26 13:15:35.000000000","tz":0},"subject":"Identity providers can be created with specified domain","message":"Identity providers can be created with specified domain\n\nIDPs are linked to a keystone domain. Currently when we provision a new IDP\nit gets created within a new autogenerated domain. Provisioners can now\ngive a domain_id in an IDP entry within the trusted_idp_list and the IDP\nwill be created within that domain.\n\nDepends-On: If761dd4faf63cf09d436d9e0a22f7e90cec9a105\nChange-Id: I30bcbce4c75b38f96a27ee742d15cf2abe402176\n\nAdd IDP display_name to defaults\n\nAllows operator a name different to the one used in the backend\nto display to the user in the horizon Identity Provider dropdown.\n\nChange-Id: I79a32f696f592b6fb3fd2c01dc9f077597b1a7a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ca7002b79b7041e85b57a92e8e53280a08bebc03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ca7002b79b7041e85b57a92e8e53280a08bebc03"}]},"branch":"refs/heads/stable/train"},"97d17b821e9b60d8b72e4d32af1b2ec58aff149d":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-06-29 11:39:45.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/02/738202/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/02/738202/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/738202/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/738202/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/738202/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/02/738202/3"}}},"commit":{"parents":[{"commit":"579fd7bed69548884bf92aed3ba60eaab9cdb92a","subject":"Add memcached caching","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/579fd7bed69548884bf92aed3ba60eaab9cdb92a"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-06-15 18:52:21.000000000","tz":60},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-06-29 11:39:42.000000000","tz":0},"subject":"Identity providers can be created with specified domain","message":"Identity providers can be created with specified domain\n\nIDPs are linked to a keystone domain. Currently when we provision a new IDP\nit gets created within a new autogenerated domain. Provisioners can now\ngive a domain_id in an IDP entry within the trusted_idp_list and the IDP\nwill be created within that domain.\n\nDepends-On: If761dd4faf63cf09d436d9e0a22f7e90cec9a105\n\nAdd IDP display_name to defaults\n\nAllows operator a name different to the one used in the backend\nto display to the user in the horizon Identity Provider dropdown.\n\nChange-Id: I79a32f696f592b6fb3fd2c01dc9f077597b1a7a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/97d17b821e9b60d8b72e4d32af1b2ec58aff149d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/97d17b821e9b60d8b72e4d32af1b2ec58aff149d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
