)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~739054","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~Iaf9f1b9198f14c903f9801e0ce7da86b74d9c5bd","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"idp-improvements","hashtags":[],"change_id":"Iaf9f1b9198f14c903f9801e0ce7da86b74d9c5bd","subject":"Identity Providers support improvments","status":"MERGED","created":"2020-07-02 14:33:33.000000000","updated":"2020-07-06 18:22:31.000000000","submitted":"2020-07-06 18:08:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739054-1594058939692-f3086ce0","meta_rev_id":"a29d83022a8fe510acca64838375e4d7a8eef933","_number":739054,"virtual_id_number":739054,"owner":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-06 18:08:59.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-07-06 15:57:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-06 16:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-06 16:11:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-06 15:57:19.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-07-06 16:11:45.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-07-06 18:08:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1b19a96f642d76daa7f0d4d9257dd713ef79172c","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-07-02 14:33:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"935c7cbafa85b1da386ebb58b7f720768a021160","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-07-02 14:40:14.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3acb3e167092bab085d832f70814d6e8723c139e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-02 17:51:08.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6098ce9e99fd4f268afd59b5844ed788 : SUCCESS in 10m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0a3e6360747147d183b4aff7c87e7e7e : FAILURE in 49m 17s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/28a5b81d5c2c441593aeaa9f9e906383 : SUCCESS in 1h 44m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/61be5ae49cfd4163a463ad063ae34351 : FAILURE in 39m 34s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/dc603a3cf9614495839ba7da286f717b : SUCCESS in 1h 10m 11s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4df8c5014fb443d3bc565027fb0b7432 : FAILURE in 1h 35m 46s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/95a691ea1a9f4c0ab8bc040e05f04246 : SUCCESS in 1h 57m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2f678cd4c0de4285bcd9d33b1a01a996 : SUCCESS in 1h 02m 31s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/d452cc35790a4f1c8dae6050f4753555 : TIMED_OUT in 3h 04m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60dc3fefbf024c8cbeea6f53089bb9e7 : SUCCESS in 10m 27s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/82feb8248c6547fea1f65b81e401028e : SUCCESS in 30m 46s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a0cc5540fad54e869a5c26ccfe90cf0d : SUCCESS in 54m 47s","accounts_in_message":[],"_revision_number":2},{"id":"72846b31538f7de467d281a051b321d38df50139","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-06 13:06:41.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"6fa1c96e5fc33f6d73805db92fe9ae54fdd573f9","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-06 13:08:53.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1a9ce8d23e1a5e8a8f4207a05d08abf202562a38","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-06 15:57:19.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"bac57c2177316f0262b8e36bfe3391173ffd8f69","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:08:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4f102c19e8744b039e407b5f473a8895 : SUCCESS in 8m 33s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/9330519abc8c4842b951dfa2c31dbb6d : SUCCESS in 1h 13m 48s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/c80d3219995d4d89b33ed48918d0ab49 : SUCCESS in 1h 34m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7b4787a1f5db475e8a2ed0c6d489dbfa : SUCCESS in 1h 12m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1df4099357d643aeb175c75dfb604a8f : SUCCESS in 1h 04m 31s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/629d58d402f14648903176afcabf050d : SUCCESS in 2h 52m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/575e9a868bca4bbe9022f7969b5a342b : SUCCESS in 1h 56m 16s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fca21ead266c471a9a0655233d1129bf : SUCCESS in 1h 20m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12c5a77b475846a4975b4bed9829ecc3 : SUCCESS in 1h 07m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61aba41037bd417186ddb69dc1661590 : SUCCESS in 10m 39s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/a5d919b2c4084ea385fb6a70573eab4b : SUCCESS in 46m 26s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5ccd88f72cb540109b843a9bb9c2edae : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":3},{"id":"b0faf3f67c81a0eb3a607a6e5e5e9d0a09671ccb","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-06 16:11:45.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"1febd7c0c503c8fd1934667811e2dd66ee9fea3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 16:11:57.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"59199d84a433e12497fc0c53252599f14fa8a99c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 18:08:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c46189b6f36e29e836d93171007621673dfe4b29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 18:08:59.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/d799b0559bfd4343b515b60b29328199 : SUCCESS in 10m 28s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/54e5fd45e3204b58896227ef4ac811e4 : SUCCESS in 1h 03m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/107ee72db4b843088ed74be36ff228a7 : SUCCESS in 1h 34m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/17694e90434540ffb57d7f31cb1a0c86 : SUCCESS in 1h 22m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0289d0a7e6a944e08078ea39e9505318 : SUCCESS in 1h 18m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/467567834fff4deb939b3a54fd3388fb : SUCCESS in 1h 01m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/83a868c9609d4c42a1a10a82f5b36c33 : SUCCESS in 1h 04m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/585748550ee649a7aa181d81a1e05df4 : SUCCESS in 11m 38s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/7589237bb2a841ea833c1f3838cdc930 : SUCCESS in 46m 45s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/07c125211a714e78a063a0af7705c093 : SUCCESS in 37m 11s","accounts_in_message":[],"_revision_number":3},{"id":"a5565991fb51afebdd319596ae7a47371cb89c45","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 18:12:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33d42873376a4033b135d0c58be8a98c : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":3},{"id":"b7c28e60c62212aee01580fff1a5ed8683a4e7c9","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-06 18:21:44.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 9e5742f7222104aefc7be00edbb60ed9124d105e","accounts_in_message":[],"_revision_number":3},{"id":"76cc5854bb2bb5f4ab4794246f7070cacf3bd262","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-06 18:22:31.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 44b7ca7d3f8f53a6e2434cfaff103c5c647abfb0","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4c9642765e632d409c6d9028aae26f4514af7033","revisions":{"e5d702c04566885f04239f4c033b62a3be33be9e":{"kind":"REWORK","_number":1,"created":"2020-07-02 14:33:33.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/54/739054/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/54/739054/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/739054/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/739054/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/739054/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/739054/1"}}},"commit":{"parents":[{"commit":"7199dcd543f2e4001def7ce2836acf6b9edc14a9","subject":"Merge \"Add OIDCAuthRequestParams parameter to template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7199dcd543f2e4001def7ce2836acf6b9edc14a9"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-06-15 18:52:21.000000000","tz":60},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-07-02 14:31:12.000000000","tz":60},"subject":"Identity Providers support improvments","message":"Identity Providers support improvments\n\nIdentity providers can be created with specified domain\nIDPs are linked to a keystone domain. Currently when we provision a new IDP\nit gets created within a new autogenerated domain. Provisioners can now\ngive a domain_id in an IDP entry within the trusted_idp_list and the IDP\nwill be created within that domain.\n\nAdd IDP display_name to defaults\nAllows operator a name different to the one used in the backend\nto display to the user in the horizon Identity Provider dropdown.\n\nDepends-On: https://review.opendev.org/#/c/736739/\nDepends-On: https://review.opendev.org/734672\nDepends-On: https://review.opendev.org/735289\nDepends-On: If761dd4faf63cf09d436d9e0a22f7e90cec9a105\nChange-Id: Iaf9f1b9198f14c903f9801e0ce7da86b74d9c5bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e5d702c04566885f04239f4c033b62a3be33be9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e5d702c04566885f04239f4c033b62a3be33be9e"}]},"branch":"refs/heads/master"},"e365eed1e7ea8598b1ebe215ebd860d906b352e7":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-07-02 14:40:14.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/54/739054/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/54/739054/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/739054/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/739054/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/739054/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/739054/2"}}},"commit":{"parents":[{"commit":"7199dcd543f2e4001def7ce2836acf6b9edc14a9","subject":"Merge \"Add OIDCAuthRequestParams parameter to template\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7199dcd543f2e4001def7ce2836acf6b9edc14a9"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-06-15 18:52:21.000000000","tz":60},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-07-02 14:40:07.000000000","tz":0},"subject":"Identity Providers support improvments","message":"Identity Providers support improvments\n\nIdentity providers can be created with specified domain\nIDPs are linked to a keystone domain. Currently when we provision a new IDP\nit gets created within a new autogenerated domain. Provisioners can now\ngive a domain_id in an IDP entry within the trusted_idp_list and the IDP\nwill be created within that domain.\n\nAdd IDP display_name to defaults\nAllows operator a name different to the one used in the backend\nto display to the user in the horizon Identity Provider dropdown.\n\nChange-Id: Iaf9f1b9198f14c903f9801e0ce7da86b74d9c5bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e365eed1e7ea8598b1ebe215ebd860d906b352e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e365eed1e7ea8598b1ebe215ebd860d906b352e7"}]},"branch":"refs/heads/master"},"4c9642765e632d409c6d9028aae26f4514af7033":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-07-06 13:08:53.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/54/739054/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/54/739054/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/739054/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/739054/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/739054/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/54/739054/3"}}},"commit":{"parents":[{"commit":"c9db38b60a3b13195bfeb7b76ac8510572c4f816","subject":"Merge \"Add Centos-8 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c9db38b60a3b13195bfeb7b76ac8510572c4f816"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-06-15 18:52:21.000000000","tz":60},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-07-06 13:08:53.000000000","tz":0},"subject":"Identity Providers support improvments","message":"Identity Providers support improvments\n\nIdentity providers can be created with specified domain\nIDPs are linked to a keystone domain. Currently when we provision a new IDP\nit gets created within a new autogenerated domain. Provisioners can now\ngive a domain_id in an IDP entry within the trusted_idp_list and the IDP\nwill be created within that domain.\n\nAdd IDP display_name to defaults\nAllows operator a name different to the one used in the backend\nto display to the user in the horizon Identity Provider dropdown.\n\nChange-Id: Iaf9f1b9198f14c903f9801e0ce7da86b74d9c5bd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/4c9642765e632d409c6d9028aae26f4514af7033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/4c9642765e632d409c6d9028aae26f4514af7033"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
