)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~739570","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fussuri~Iaf9f1b9198f14c903f9801e0ce7da86b74d9c5bd","project":"openstack/openstack-ansible-os_keystone","branch":"stable/ussuri","topic":"idp-improvements-stable/ussuri","hashtags":[],"change_id":"Iaf9f1b9198f14c903f9801e0ce7da86b74d9c5bd","subject":"Identity Providers support improvments","status":"MERGED","created":"2020-07-06 18:21:44.000000000","updated":"2020-07-13 16:04:20.000000000","submitted":"2020-07-13 16:01:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"739570-1594656087747-eb87a29e","meta_rev_id":"7446e8cc4d7a55774feddf6defca338720b7b6cb","_number":739570,"virtual_id_number":739570,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-13 16:01:27.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":2,"date":"2020-07-13 14:13:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-06 21:59:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"all":[{"value":1,"date":"2020-07-13 14:13:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-13 14:13:21.000000000","updated_by":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"reviewer":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"state":"REVIEWER"},{"updated":"2020-07-13 16:01:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"450390233a991c1066f5de588482c06bf7a4c316","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-06 18:21:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"10058352a7d87dd44c519f9c0d0a8e67094c0129","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-06 21:05:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4ac642df07364ddc9be839ccde754b45 : SUCCESS in 8m 00s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ca87fec826b24b35b59501b41a20b84b : SUCCESS in 1h 31m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d753eed7fa8e4e26812295ba7620d577 : SUCCESS in 1h 33m 33s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7401017dbe644e82ab88a132a890acca : SUCCESS in 1h 16m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/707edf236ee842c18ac0523f54707ada : SUCCESS in 1h 51m 01s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/dadeea5100e74fe7b81f6f54978d8256 : SUCCESS in 2h 34m 42s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/97f1c4655e4f43d4ac14e4b9e312f406 : SUCCESS in 2h 05m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0f365b6eaccb4f78bbc2748efbe73415 : SUCCESS in 56m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2883dc6bb1ad4e639892a6555ce7d243 : SUCCESS in 1h 18m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90bd6fe0ebe44d67b747ac8c0a06a89b : SUCCESS in 9m 07s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/a7551ca75c0142a7a9332a6840818f92 : SUCCESS in 44m 02s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e4171754d3484f8ea897003af4f49231 : SUCCESS in 37m 15s","accounts_in_message":[],"_revision_number":1},{"id":"c802c4d1595f321a6f3ebaa7a3aa8a5b2b0cb406","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-06 21:59:43.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b1bc83777c2a4ff96178bd7cec6d56852c0fd82c","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2020-07-13 14:13:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8e817fff7811c92b5449831078282aedcab21e58","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 14:13:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"74037f075afc85b01d79e235e7a567fe5cbede17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 16:01:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9d499c8601842d2502292553c1a7c613a2cde48e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 16:01:27.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0af1d122cc17451eb40633859f96da0f : SUCCESS in 13m 10s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/71f39bf92fe5477aa9cac16c903302b5 : SUCCESS in 49m 35s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/27f7858f7a0d49788ef9ddd9572e7a96 : SUCCESS in 1h 44m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99be4b264ecc4873badf09bf7e7f6e75 : SUCCESS in 1h 31m 48s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e7722c4e0679454d8ae0c2b39f0ced9f : SUCCESS in 1h 32m 32s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e5a1f63231384d3eb3d71dfef2cd0d99 : SUCCESS in 1h 05m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cff4e7fca8d44d71b846f0c54584c83d : SUCCESS in 56m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9a7236b03ce45c784455243b2608469 : SUCCESS in 6m 00s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/ec812167279149e68200d2acf86c036c : SUCCESS in 39m 17s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/515b71581d854c9d9c26daeeee5edae8 : SUCCESS in 35m 18s","accounts_in_message":[],"_revision_number":1},{"id":"da7a35b06cd891e74bb71e9e929f259de1437581","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 16:04:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34578f034f224f789695c37a76134fb6 : SUCCESS in 2m 30s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e5742f7222104aefc7be00edbb60ed9124d105e","revisions":{"9e5742f7222104aefc7be00edbb60ed9124d105e":{"kind":"REWORK","_number":1,"created":"2020-07-06 18:21:44.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/70/739570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/70/739570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/70/739570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/70/739570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/70/739570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/70/739570/1"}}},"commit":{"parents":[{"commit":"c306cdc84465f793999f81802acafb754aab8819","subject":"Merge \"Add OIDCAuthRequestParams parameter to template\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/c306cdc84465f793999f81802acafb754aab8819"}]}],"author":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2020-06-15 18:52:21.000000000","tz":60},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-06 18:21:44.000000000","tz":0},"subject":"Identity Providers support improvments","message":"Identity Providers support improvments\n\nIdentity providers can be created with specified domain\nIDPs are linked to a keystone domain. Currently when we provision a new IDP\nit gets created within a new autogenerated domain. Provisioners can now\ngive a domain_id in an IDP entry within the trusted_idp_list and the IDP\nwill be created within that domain.\n\nAdd IDP display_name to defaults\nAllows operator a name different to the one used in the backend\nto display to the user in the horizon Identity Provider dropdown.\n\nChange-Id: Iaf9f1b9198f14c903f9801e0ce7da86b74d9c5bd\n(cherry picked from commit 4c9642765e632d409c6d9028aae26f4514af7033)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9e5742f7222104aefc7be00edbb60ed9124d105e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9e5742f7222104aefc7be00edbb60ed9124d105e"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
