)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~742075","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fussuri~I018d1fc8ad637de8e76a98f22f6cef21e3f0b7c3","project":"openstack/openstack-ansible-os_keystone","branch":"stable/ussuri","topic":"osa/el8-stable/ussuri","hashtags":[],"change_id":"I018d1fc8ad637de8e76a98f22f6cef21e3f0b7c3","subject":"Add Centos-8 support","status":"MERGED","created":"2020-07-21 06:38:04.000000000","updated":"2020-07-21 14:51:25.000000000","submitted":"2020-07-21 14:48:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"742075-1595342922681-da4fe202","meta_rev_id":"8193abd5ae6a9c4da0c9a1e04f39293c3d6dcbf7","_number":742075,"virtual_id_number":742075,"owner":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-21 14:48:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-07-21 08:21:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-21 08:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-21 08:56:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-21 08:21:16.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-07-21 14:48:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"166f30b2ecbeb1b03cc7c76cbafb16bf38209764","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-21 06:38:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ed8c67d91a61b7498c932cdc3e2599e42e08cdf1","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-07-21 08:21:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b6fd5a9b583d94f0d5edfd5fbc91e973d3f972e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 08:54:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/f1773c4a4ac24e389e7c10efc8c6f38f : SUCCESS in 8m 06s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/16f0a5a9d2824dc58bc31314b1bf6600 : SUCCESS in 1h 15m 41s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/032ceb37bc8846cfb86644b502d8f519 : SUCCESS in 1h 21m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5502d94ffd59420d85726f2ada91d663 : SUCCESS in 1h 18m 21s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91771e5ce05a45ef8a9ec750f0e7618f : SUCCESS in 1h 23m 26s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3f37c9fa67b44528885d9da931c28fbd : FAILURE in 1h 46m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b72248b4710047aab9c8ac37a7276ca6 : SUCCESS in 1h 59m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/01fd26a3bd9d4760a09d54ceee8996a2 : SUCCESS in 1h 11m 04s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/7d99e417d468485db342bbf3db0731b1 : SUCCESS in 1h 02m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09c9d0b49d7240b794318109dae35d6a : SUCCESS in 9m 39s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/182896a693e9455aa6b476d987f7c21f : SUCCESS in 39m 55s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c467b01c2d004da78907e9111036a347 : SUCCESS in 37m 21s","accounts_in_message":[],"_revision_number":1},{"id":"f47f6b6989b3664da9d7feab60d3d8d0ffa26360","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-21 08:56:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"298652602a10384cf9910292dd10b781afcd5e68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 08:57:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c9a73b3e1494558bd77b5533f92cce626f5ac52","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 10:43:35.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/b98ac0110d1b4d568b92647f9989a2f5 : SUCCESS in 7m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d7cd85c9d00e41b4b938e870c3510e36 : SUCCESS in 1h 08m 44s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/72af78753530476c830aa666287e11e5 : SUCCESS in 1h 34m 08s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fc6df4a89794f4691d8a947ad1862e7 : SUCCESS in 1h 20m 43s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2cef47e3cd346639cc79dee19a9f92f : SUCCESS in 1h 12m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/750d9a922c8a4a2b81d9d7eebd459894 : SUCCESS in 1h 05m 15s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c9d9da4819d34e9dbbb644696ed6044c : SUCCESS in 58m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b12d0dc84a04ee182632f9c1d941f77 : POST_FAILURE in 16m 01s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/209a2f415e8247f1bbb66df1df3a129d : SUCCESS in 42m 00s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/18beaaa0c8e94401a2d3492508c521ca : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":1},{"id":"87daa3ffdbf321e327aed9bb20c98cc9fdeb1a88","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-07-21 10:58:19.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b5d08d3e44be92d9ddf755ab9e0e8593be1e2b9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 13:07:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/41abbf916a5f472f95ed1ebb2ac3ca08 : SUCCESS in 9m 41s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/60bf7cc8ebd446b0bc3ee561c1959a73 : SUCCESS in 1h 28m 53s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6bcd8b0fbf304910968c265ed515204a : SUCCESS in 1h 18m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/72023cc7bbc74defb3b2e8cc6f82cbcc : SUCCESS in 1h 09m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/29b718dcf47d4544b1a636fa2cff9dc8 : SUCCESS in 2h 01m 01s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/9c33bb9cae194aae9c1da450badcd954 : FAILURE in 1h 26m 28s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/706a64fe66e84b56a8e02e9d47969c08 : SUCCESS in 1h 57m 07s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/753509c097be4818b4429325bbc3ac54 : SUCCESS in 1h 02m 27s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1b085ab4b37f4ae3b4b5d5ea0f8022d8 : SUCCESS in 55m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26dcb7aae7c248f3a58f5d4c2c3c60ec : SUCCESS in 8m 33s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/e7d6779cbf3649bfa8096d129416ccd4 : SUCCESS in 44m 26s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6cc2b1591c1b4d0c8aaac2b59aa1174d : SUCCESS in 39m 39s","accounts_in_message":[],"_revision_number":1},{"id":"03aeb82613bf2e7d75687b2968512e5d7a718ea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 13:07:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a3d479320f67359938949d533f098aff48577253","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 14:48:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/21b19e6efbe64bfd8de095b8eeee78da : SUCCESS in 8m 53s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ad0b421ef5b24dea9ffcc98898b47e47 : SUCCESS in 1h 08m 36s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bb49351d448c4779b72c491329cb409e : SUCCESS in 1h 28m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/c336f7b476b349a080c9df00732e6d2c : SUCCESS in 1h 14m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54076918e27a4af78e2f0d5c25611812 : SUCCESS in 1h 04m 19s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/15e3bee8e8f0462faf60c5e8af2993d5 : SUCCESS in 1h 03m 00s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/45a523fc12b440d69f28776650eb7d6f : SUCCESS in 58m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69dc7a32c5e24360a99b29d0c3a242c9 : SUCCESS in 6m 56s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/592db9102b13411983465cfe8fb95a3a : SUCCESS in 40m 15s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/42c760f649a340449b0f95a629993277 : SUCCESS in 37m 02s","accounts_in_message":[],"_revision_number":1},{"id":"9c7ec728bb444a7e6226bc7750ba77d5d7c929d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 14:48:42.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cf9ce746446535477a7adbbce5c644be662ddbd8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-21 14:51:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88cccb2d5e924ad78eb0a99ca3e0bc66 : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dcf1da03a17070a0eb68b84fe4515ba35107a806","revisions":{"dcf1da03a17070a0eb68b84fe4515ba35107a806":{"kind":"REWORK","_number":1,"created":"2020-07-21 06:38:04.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/75/742075/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/75/742075/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/75/742075/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/75/742075/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/75/742075/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/75/742075/1"}}},"commit":{"parents":[{"commit":"9e5742f7222104aefc7be00edbb60ed9124d105e","subject":"Identity Providers support improvments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9e5742f7222104aefc7be00edbb60ed9124d105e"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-06-15 10:58:52.000000000","tz":180},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2020-07-21 06:38:04.000000000","tz":0},"subject":"Add Centos-8 support","message":"Add Centos-8 support\n\nChange-Id: I018d1fc8ad637de8e76a98f22f6cef21e3f0b7c3\n(cherry picked from commit be3a7c7e2bf63a73a1b971141e5f511bce2be5a9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/dcf1da03a17070a0eb68b84fe4515ba35107a806"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/dcf1da03a17070a0eb68b84fe4515ba35107a806"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
