)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~745850","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I8e306eaea3ee184044e7fd9655a57ca9f5026c6b","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"openstack/openstack-ansible-tests/sync-tests","hashtags":[],"change_id":"I8e306eaea3ee184044e7fd9655a57ca9f5026c6b","subject":"Updated from OpenStack Ansible Tests","status":"MERGED","created":"2020-08-12 11:33:30.000000000","updated":"2020-08-15 23:54:40.000000000","submitted":"2020-08-15 23:53:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"745850-1597535613378-b7eb80e6","meta_rev_id":"1047df3027960eb559f170af57fab8a2fc671aab","_number":745850,"virtual_id_number":745850,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2020-08-14 22:42:21.000000000","_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-15 23:53:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-13 05:41:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-13 05:41:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-13 05:41:11.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-08-14 22:42:21.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"state":"REVIEWER"},{"updated":"2020-08-15 23:53:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b03516b4bac49dfce0ad10633d5004d8b5b0eae6","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2020-08-12 11:33:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f463a5772b132f6648e4241d189bb5b17e9e536d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-12 14:13:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/670688c3423d493f90844ba173148722 : SUCCESS in 6m 24s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/530afccc2f104923bb5c2630ed72e817 : SUCCESS in 1h 19m 04s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/79a2ce9b41b94831adf6a3d82c4c8ff5 : SUCCESS in 1h 14m 56s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/feb3c7e78713426584b2b79b60e35ab8 : SUCCESS in 1h 18m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/0718be0da0544505aabd48990afd5285 : SUCCESS in 1h 16m 00s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56a5f86e5d174c7dbfcf36c34a1bed2d : SUCCESS in 1h 01m 38s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/8086181cd22c4f7ab64541c8c6c383cf : SUCCESS in 2h 16m 51s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1928adb2e30245d5851bfb4fd0e79a4b : SUCCESS in 1h 52m 35s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/0a29794f3db64832b5204ab022ffa917 : SUCCESS in 1h 11m 40s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/42c394d07eef472ab66acadb655111f3 : SUCCESS in 53m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6fd7c9e9aa3a4158b0c34ee1d3979077 : SUCCESS in 1h 16m 40s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f1512b4c768c4f48af8e2e72df83b743 : SUCCESS in 53m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57697d2952e74429aafd15fd1c44714f : SUCCESS in 5m 18s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/57275fd130bc4d62bf0ec584d3ef5aeb : SUCCESS in 34m 08s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2c790817f0fb45d9bf2192ddcbb4f4bb : SUCCESS in 32m 12s","accounts_in_message":[],"_revision_number":1},{"id":"c36514cde83d2476007b52de53aef1300728f55a","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-13 05:41:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1d1e28e9842977833712295fd59875e358b75c34","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 05:41:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ca0cd7aa0f10cc6810c2f931bed091b28974826d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-13 08:09:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/a3e81a747d4f4281a89c766bade53c5e : SUCCESS in 7m 50s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/86627a83fbbd431fbef72e483d3098b5 : FAILURE in 21m 21s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/e0815930a9604eba9db993e8cfa323b7 : SUCCESS in 1h 16m 17s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/339750a7294f4b18902b947d88da8ced : SUCCESS in 1h 16m 24s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/518edb060d0d4991acbf6558ff15bb0f : SUCCESS in 1h 16m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/3b26790c8fca433c8cd0c4599098b228 : RETRY_LIMIT in 8m 44s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/5fe9ccef75844885b0ab5de020c90038 : SUCCESS in 1h 08m 11s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6a9e81d730e9414e9c83bc617b2b3847 : SUCCESS in 1h 10m 25s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/3adbadb4ed9542c8b3c2c262cdc62ac6 : SUCCESS in 1h 11m 02s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/df7d99e4a3104786af11a468568cd3e1 : SUCCESS in 54m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9f6d8027b842b294f1f9d7713fb1c8 : SUCCESS in 6m 36s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/cb55e99b2a154115a276db4e8cc74dfe : SUCCESS in 54m 20s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/739a1b67a1a94c88932fcac9191f43a5 : FAILURE in 35m 39s","accounts_in_message":[],"_revision_number":1},{"id":"b29c3e8f0fcf63618082849265cefe11397cbeaa","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-14 21:10:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c30d32e2d7f57c2a53858727ff2fa54f903ab7f4","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-08-14 22:42:21.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"26bfc8682f8f1103ac36d2ae577d269ea4a7d0ae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-14 23:33:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/4e5e3107c12541e5b8b7522212d72dad : SUCCESS in 5m 36s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/ec87d2377c134fdfbb289ea750958eac : SUCCESS in 1h 03m 59s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/956fe53a81d34187be104168cd1fda79 : SUCCESS in 55m 09s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/fbee0221b0f6497496339f9198822335 : SUCCESS in 1h 10m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cf9a98e63c7640e79a3b131f697cb426 : SUCCESS in 1h 09m 28s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/24d602b6c1d849aa8a2887ec1a12cee9 : SUCCESS in 56m 45s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/bf5b9c0fdb5c4592a776d67657147611 : SUCCESS in 2h 15m 58s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/6f11f2d85d774060806cc5fb443d1ec3 : SUCCESS in 1h 43m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/3ca6ab19981e41a8a7bf3c50a8083072 : SUCCESS in 1h 01m 59s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/29fb7648ca1a40629f707f7665ce539e : SUCCESS in 59m 42s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/81365c803e6a40b19f3a2afcf15ab2a1 : SUCCESS in 49m 17s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/419823c702de482d89ddcc19cde73291 : SUCCESS in 54m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5f0fd4b532ec4f94a91df898918ef326 : SUCCESS in 5m 03s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/4022ab7af35a405fae435188ed82384f : FAILURE in 4m 41s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fbba17496a3949dda2291c156dcd0f14 : SUCCESS in 32m 01s","accounts_in_message":[],"_revision_number":1},{"id":"762d39f3febcb06c4accbdb92f29894df9112ab1","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-15 14:19:24.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"c1ecd3c042ec3f2cfb1dd184b8a851a99e5ce392","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 16:57:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/25090428506f46be9b76a02df183c283 : SUCCESS in 6m 44s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/addffaac4f064429b740a3bbde6503e2 : SUCCESS in 1h 03m 55s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/7416ba6f268246d095afedacee31038c : FAILURE in 23m 18s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/99a819a2c8e8490b81e03fb7afe5fc7e : SUCCESS in 1h 13m 45s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2e28757590a248da8a63f4c8a354932b : SUCCESS in 1h 06m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1ce935d459d84e23994a891922b2c0fa : SUCCESS in 1h 00m 29s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/cae2b486cdac497ba6b5782a77e7e6cd : SUCCESS in 2h 21m 21s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/534dce1336cd45b1a3d27f557c4e0744 : SUCCESS in 1h 38m 42s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6c476ed688694a5fa11e45465f58e5f9 : SUCCESS in 1h 42m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0a76bea974984ccc8e72be768e461de5 : FAILURE in 1h 03m 51s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19a97a1a19e84253b0a4b62347f5b394 : SUCCESS in 54m 48s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b69cfc8a629248f2b62f88776d6f9d20 : SUCCESS in 49m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f905609255eb47db90fc446eafa833c7 : SUCCESS in 5m 28s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/4aa1cccea8584df1bcbcbe46967f7559 : SUCCESS in 33m 43s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/fa60a0a09d2f490595951256d7173b1a : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":1},{"id":"947a01c453a3643bc171a5e4ff94c015a19df119","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-08-15 20:22:55.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"441e629dbd325a638aca31782a1079921dbd21d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 22:23:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0e64006158354d99a46dd464993bce04 : SUCCESS in 8m 30s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/745b691ec406441c94b96260a23e0dd4 : SUCCESS in 1h 09m 01s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/95fcc026dd74460eb1d32eeac4f17b0a : SUCCESS in 1h 23m 53s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e5e3255dc4ec41a9800cb6637e323932 : SUCCESS in 1h 20m 12s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/62f0221c4a6f423abbe877afcea065e4 : SUCCESS in 1h 05m 15s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0be94b0bf9d347f8a88e66089b78db1b : SUCCESS in 1h 08m 23s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/fc248362b763410785d2a446d2c040eb : SUCCESS in 1h 56m 23s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/69bc3e0497274ec4b3b6890ec84f7973 : SUCCESS in 1h 44m 49s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/4546604589dc4ba5a809e62be7806e42 : SUCCESS in 56m 22s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/d4b2285eb6f74c818bd718f0950cfe3c : SUCCESS in 1h 04m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/21a4a426a4ed4e4a873dfb4b2181a704 : SUCCESS in 53m 46s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/32bd724132254bdc96bc6f29f596c639 : SUCCESS in 56m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66dfefcf86294241a08dd416279555c9 : SUCCESS in 5m 23s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/28658507e3d94463afec802425d0cecc : SUCCESS in 32m 08s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/66b2e6c05b2e44f79155d4e449835453 : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":1},{"id":"de740dd1354a581bd0d88b8ee64742800d9bfd00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 22:23:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"db73476a2f2c7a52b9faff320fedf06e91325f25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 23:53:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/471f9b1ce30444fab81ffb265d9f55c5 : SUCCESS in 6m 20s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6a4593bcd8cf48399f8392b8296ca140 : SUCCESS in 1h 10m 37s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/2afc6b154be3448f80021e9df4de65e6 : SUCCESS in 1h 23m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/18810df7ffd044d2a184c763150f98ef : SUCCESS in 1h 09m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f735a4bdd2ca40c3987839f9c7998e2f : SUCCESS in 1h 12m 46s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e0d9e606f36d4da99e0e3279b79e095e : SUCCESS in 1h 01m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6855683f43ed4c36bc2591a7613d2261 : SUCCESS in 58m 50s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e298d4d82fe842a18a85d58f73d8de2c : SUCCESS in 56m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1bb9a179cbd34dbd8c41781703c74a4a : SUCCESS in 52m 49s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/94a18b18c0d24c59939698fe655d6ee2 : SUCCESS in 45m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/588561b2df94468a9913642ddf13c24d : SUCCESS in 4m 45s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/b94177034b834a92aa92f5730f15e052 : SUCCESS in 32m 29s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b01b439ee9a649b89bdd4f2f0f8627a5 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":1},{"id":"1505646a9e743f3d3c1c51d733b4b481ea304aa4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 23:53:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7c604cff5c024bbb6b11490769f7bea750e28d9f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-15 23:54:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0edc9bece84641b99e1ecd5a3980fd36 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9447a108207115dab35c4d196f36b90c5cad2381","revisions":{"9447a108207115dab35c4d196f36b90c5cad2381":{"kind":"REWORK","_number":1,"created":"2020-08-12 11:33:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/50/745850/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/50/745850/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/50/745850/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/50/745850/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/50/745850/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/50/745850/1"}}},"commit":{"parents":[{"commit":"4c9642765e632d409c6d9028aae26f4514af7033","subject":"Identity Providers support improvments","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/4c9642765e632d409c6d9028aae26f4514af7033"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-12 11:33:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2020-08-12 11:33:28.000000000","tz":0},"subject":"Updated from OpenStack Ansible Tests","message":"Updated from OpenStack Ansible Tests\n\nChange-Id: I8e306eaea3ee184044e7fd9655a57ca9f5026c6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9447a108207115dab35c4d196f36b90c5cad2381"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/9447a108207115dab35c4d196f36b90c5cad2381"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
