)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~752310","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Ftrain~I43df6a6b70195e21e0845565d12aeed5a34d2cda","project":"openstack/openstack-ansible-os_keystone","branch":"stable/train","topic":"libssl_dev-stable/train","hashtags":[],"change_id":"I43df6a6b70195e21e0845565d12aeed5a34d2cda","subject":"Add libssl devel package for CentOS and Suse","status":"MERGED","created":"2020-09-16 20:25:21.000000000","updated":"2020-09-18 09:01:20.000000000","submitted":"2020-09-18 08:57:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"752310-1600419465245-dddb1e4e","meta_rev_id":"3b63d7b029d73eb671fa646e61e66ae241cb9ca9","_number":752310,"virtual_id_number":752310,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-18 08:57:45.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":2,"date":"2020-09-18 07:02:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-18 06:41:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":1,"date":"2020-09-18 07:02:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-18 06:41:02.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2020-09-18 08:57:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3ddd8290b42f0e9a89a868e4b6a63bea3123c89a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-16 20:25:21.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0b6e3d21c67d792bae2c56326535301d93e4191f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-16 23:05:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/0280f7f338c44f278c15aec59af9dda8 : SUCCESS in 8m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6e102af6f08e4ad5bbcd9695b55dfe0f : SUCCESS in 58m 05s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/bfb5d11faca143178eac7abdfc754331 : SUCCESS in 1h 16m 03s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/206d10f180414d11bd5dc2e388292b7d : SUCCESS in 1h 25m 44s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/f86d721e1a83412d8cbb95044b1ab782 : RETRY_LIMIT in 27m 26s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/10686f097b0045859655c8782c0e3189 : SUCCESS in 1h 03m 17s\n- openstack-ansible-upgrade-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/6caa7aebd33e4ec2bedf2fd00b92ebb9 : SUCCESS in 2h 29m 06s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/4507645814714e2f80132293c96bb07b : SUCCESS in 1h 47m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/b8c93ab895d249e09c16e18e6c809179 : SUCCESS in 1h 05m 18s\n- openstack-ansible-deploy-aio_distro_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/99f9926e441043f5bdef49f76e91aea5 : RETRY_LIMIT in 28m 10s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/953be707bd9240c981a47bdd00765367 : SUCCESS in 52m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3351824be3814d63ab93fadfe035bd03 : SUCCESS in 14m 56s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/09e742ae59ee4206a9e2e4a4fdebf6f4 : SUCCESS in 39m 10s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1f882222fcc44f68a17edcd4096c165b : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":1},{"id":"75abba45714bc5ad23210b97c2552b0ab94ad6f7","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2020-09-18 06:41:02.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0c859a3dde850834ce3707eef074b87d1f68545a","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2020-09-18 07:02:50.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a128e3ea5e137e683798be8393cb9c22cfe7cd68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 07:03:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c5234c22c9c296b7546024215ecdc4185f3f12d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 08:57:45.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c145365b751a0bd08dfc69032342f80c3afd9ca6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 08:57:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/97ec1787dfff4645a71264311867ecdb : SUCCESS in 9m 53s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/1b4f0101d5584aa1aa972aa950cd88c4 : SUCCESS in 54m 30s\n- openstack-ansible-deploy-aio_metal-debian-stable https://zuul.opendev.org/t/openstack/build/767c2372020d489e9a1d1777e41c7d4a : SUCCESS in 1h 19m 39s\n- openstack-ansible-deploy-aio_metal-centos-7 https://zuul.opendev.org/t/openstack/build/f3f3ccb6acdb4e5887d7f9d747e61aa1 : SUCCESS in 1h 26m 13s\n- openstack-ansible-deploy-aio_metal-opensuse-15 https://zuul.opendev.org/t/openstack/build/4fdd6a833a784cac9d1aa4bf70e86dd5 : RETRY_LIMIT in 17m 53s (non-voting)\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/09e56ea7ff9f46ff8eb58f000e6d8323 : SUCCESS in 1h 06m 24s\n- openstack-ansible-upgrade-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/a3178c7ab93a45bfa9642e30d7a55cbf : SUCCESS in 1h 42m 57s\n- openstack-ansible-deploy-aio_distro_metal-centos-7 https://zuul.opendev.org/t/openstack/build/d803c3d771714bbda31fd7d9c0748d0e : SUCCESS in 58m 33s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9534ace279544d5bb4c34987933eec21 : SUCCESS in 54m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f89873c2af94f22917449a86fe5b619 : SUCCESS in 16m 37s\n- openstack-ansible-uw_apache-centos-7 https://zuul.opendev.org/t/openstack/build/ab2cfe6f77ec4ec1ab6474a694ee9ad8 : SUCCESS in 38m 25s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/253fbf1e3e144412add4bde28a7fd9e5 : SUCCESS in 31m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f6d2d007bac7ec2ec06cab9ef11607ae63691e2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-18 09:01:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49d5ebc70e704b72b0ae100e801c65d9 : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"caedd0da32eb8c7a3d32b5c7d025232630b737f9","revisions":{"caedd0da32eb8c7a3d32b5c7d025232630b737f9":{"kind":"REWORK","_number":1,"created":"2020-09-16 20:25:21.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/10/752310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/10/752310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/10/752310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/10/752310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/10/752310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/10/752310/1"}}},"commit":{"parents":[{"commit":"6e23334a99d219f97890c514da282c41ed41ba4a","subject":"Add Paramaters to httpd.conf template","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6e23334a99d219f97890c514da282c41ed41ba4a"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"drabotyagov@vexxhost.com","date":"2020-01-28 19:01:36.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov (noonedeadpunk)","email":"noonedeadpunk@ya.ru","date":"2020-09-16 20:25:21.000000000","tz":0},"subject":"Add libssl devel package for CentOS and Suse","message":"Add libssl devel package for CentOS and Suse\n\nlibssl-devel package is required to install scrypt for keystone venv,\nwhich is direct keystone requirement [1]\n\n[1] https://opendev.org/openstack/keystone/src/branch/master/requirements.txt#L22\n\nChange-Id: I43df6a6b70195e21e0845565d12aeed5a34d2cda\n(cherry picked from commit e0197db984f8e772b9a1c235e4f347d596255603)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/caedd0da32eb8c7a3d32b5c7d025232630b737f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/caedd0da32eb8c7a3d32b5c7d025232630b737f9"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
