)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~764199","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~If3477ccc447bbd4b474b41ebfbc92e81f913aa05","project":"openstack/openstack-ansible-os_keystone","branch":"master","hashtags":[],"change_id":"If3477ccc447bbd4b474b41ebfbc92e81f913aa05","subject":"Dep\u0027s should be restricted by tox-constraints","status":"ABANDONED","created":"2020-11-25 14:39:49.000000000","updated":"2020-11-27 01:08:11.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"838b9a2e5cb41bb44ada2858ac68ab98fd372a71","_number":764199,"virtual_id_number":764199,"owner":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-11-25 15:29:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"all":[{"value":-1,"date":"2020-11-26 13:41:09.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-11-25 15:29:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-26 13:41:09.000000000","updated_by":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"reviewer":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"state":"REVIEWER"}],"messages":[{"id":"f6fa58afdb7f5f9c815d4c7c9babceac6ff28e7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-25 14:39:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff8efa13a213da9c992d1f417e045a399fa797db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-25 15:29:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/fcec948d34254214bcedb05fc70bd624 : SUCCESS in 9m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10f3b6ba5a424922ba86eb6dceec7a68 : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c72a9710ee24696ad89353b3736717d : SUCCESS in 3m 55s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/99302808d9854f2a890f454d7712665b : SUCCESS in 38m 52s","accounts_in_message":[],"_revision_number":1},{"id":"74624d8d92c1c3c88aa693b60b766de24bce4146","author":{"_account_id":32231,"name":"zhang, chun","display_name":"Shan Hai Jing","email":"zhangchun@yovole.com","username":"zhangchun"},"date":"2020-11-26 13:41:09.000000000","message":"Patch Set 1: Code-Review-1\n\nfrom the doc https://tox.readthedocs.io/en/latest/,  constraints are set in install_command will work for all test envs.","accounts_in_message":[],"_revision_number":1},{"id":"838b9a2e5cb41bb44ada2858ac68ab98fd372a71","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2020-11-27 01:08:11.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e9741491a80e0a4a833ccf13fef194f60c8a14f9","revisions":{"e9741491a80e0a4a833ccf13fef194f60c8a14f9":{"kind":"REWORK","_number":1,"created":"2020-11-25 14:39:49.000000000","uploader":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"ref":"refs/changes/99/764199/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/99/764199/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/764199/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/764199/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/764199/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/99/764199/1"}}},"commit":{"parents":[{"commit":"067669ba6f0a1e78fec4596fc93ce1509d9f6ba9","subject":"Merge \"Shibboleth and OIDC apache modules are co-installable on Focal\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/067669ba6f0a1e78fec4596fc93ce1509d9f6ba9"}]}],"author":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-11-24 15:46:06.000000000","tz":0},"committer":{"name":"wuchunyang","email":"wuchunyang@yovole.com","date":"2020-11-25 14:38:06.000000000","tz":0},"subject":"Dep\u0027s should be restricted by tox-constraints","message":"Dep\u0027s should be restricted by tox-constraints\n\nTox trying to install latest versions for building docs which may\nnot be supported by stable and lower branches, so should be\nrestricted by respective version\u0027s tox-constraints.txt\n\nChange-Id: If3477ccc447bbd4b474b41ebfbc92e81f913aa05\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e9741491a80e0a4a833ccf13fef194f60c8a14f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/e9741491a80e0a4a833ccf13fef194f60c8a14f9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
