)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~773964","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fvictoria~I68c0b138955693c8d1992f986878862ea12f5149","project":"openstack/openstack-ansible-os_keystone","branch":"stable/victoria","hashtags":[],"change_id":"I68c0b138955693c8d1992f986878862ea12f5149","subject":"Allow OIDCClaimDelimiter to be set in the apache config file","status":"MERGED","created":"2021-02-04 15:05:56.000000000","updated":"2021-02-05 18:51:31.000000000","submitted":"2021-02-05 18:50:16.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"773964","cherry_pick_of_change":773978,"cherry_pick_of_patch_set":1,"meta_rev_id":"9d568ced308f507aaf4bea35045dad3160145f89","_number":773964,"virtual_id_number":773964,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-05 18:50:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-02-05 12:22:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-02-05 11:43:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":1,"date":"2021-02-05 12:22:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-04 15:05:56.000000000","updated_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-02-04 17:13:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-05 12:22:33.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"}],"messages":[{"id":"5225995757ef623360025dd9c68b8bbd771fc94a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-04 15:05:56.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"097e2ca00e9273f8626ba7d1902d9465005e4575","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-04 17:13:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3280c87ccf564194b0234c2913a659d9 : SUCCESS in 9m 49s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b42bb70027084485913e3262e7da5a5c : SUCCESS in 1h 09m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/24e0a152c8b34f9698972d0993fe3717 : SUCCESS in 1h 26m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/707e0593509f440ea8f8207ee5d6ab8d : SUCCESS in 1h 21m 41s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/925dd194c14b4261a83a71bad2b50bb2 : SUCCESS in 1h 24m 36s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/94be41494e4849e0a5db1ae4a5d493b8 : SUCCESS in 1h 55m 30s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cccd9404f5e408997638ec92d474fcb : SUCCESS in 1h 32m 01s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9f835db9a7104e58a57992fb91996534 : SUCCESS in 1h 02m 01s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/890f8041444c4336bd6604fc9653f124 : SUCCESS in 1h 00m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f905bec3981c415f84ccc6fb65ab7e7b : SUCCESS in 4m 37s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/a80cc1607af24fc1b327ae56f7750440 : SUCCESS in 43m 43s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/214686da70d146cf8d5c885f1b4095ac : SUCCESS in 42m 35s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/216ee246a4624a01849b1d4ddff9db22 : SUCCESS in 32m 20s","accounts_in_message":[],"_revision_number":1},{"id":"55c0f1d212ae09195a6d6cf8320c7d332155b9d3","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-05 11:43:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"3b5370f09b6e4830228df9a1554c78b970700231","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-05 12:22:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8ea2392539d103b747385deea73da0500c49e602","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 12:24:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd97ef45b6e03dfd508fccb49e47be7adbd9bbe0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 13:50:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/83d95f874c4349bd8ab7f2793e6516c7 : SUCCESS in 7m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a0a88f1d6af144f8b43bc1aac0cbe909 : SUCCESS in 1h 08m 07s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/dfa8eb2ee3fe45dab7c3bfd006ffe569 : SUCCESS in 1h 03m 19s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/053ee8f468504aa1a8b736e5f37ebdba : SUCCESS in 1h 21m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/09bbc32fcd804740b0749dabaf6f3964 : FAILURE in 46m 36s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/67d508cd401f48229fed47719e9d5ce5 : SUCCESS in 1h 05m 10s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f4f0651c149b489898b5144832a47e0d : SUCCESS in 1h 00m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ed50769441343b1be7fb45f3cdaf3b1 : SUCCESS in 7m 03s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/2eb6031e5b8b4e9f8c835100ac365db8 : SUCCESS in 42m 17s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/19a4de9ec82b453aa499a809b32bd151 : SUCCESS in 37m 38s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/69738485883745bfb13ff2ea320fc589 : SUCCESS in 35m 05s","accounts_in_message":[],"_revision_number":1},{"id":"73fde4c6e7c6d731ef281dde3a3b498ae46aaf7b","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-05 15:07:12.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"58a0b29f546e2339a9cb08feefd7ade1855ec326","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 17:13:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/c79280c237494807a3ef8cceb1248c70 : SUCCESS in 8m 52s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/b553795c82ce434fbfe838fae13edb8a : SUCCESS in 1h 13m 06s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ee8111a0b180436cae9e6259e333acf8 : SUCCESS in 1h 36m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9c93993438304d639680f0af6fe41d57 : SUCCESS in 1h 28m 35s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/afe79ff995a44db690d94e2090b74683 : SUCCESS in 1h 13m 44s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/eee07aac89cc4e69a6c2fba532b7c2af : SUCCESS in 1h 53m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7321ead1370e4366aa096b8e7df52916 : SUCCESS in 1h 38m 56s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/62d5e51b1bff4cbe9aa59ee8a0870242 : SUCCESS in 1h 04m 29s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7fe7b53ebb140d2a3b88feeeb64eecb : SUCCESS in 56m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4bfb1a183df4741a532d6a1e517aeab : SUCCESS in 5m 50s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/89198262368c4fcab481f6b32f25dc78 : SUCCESS in 41m 11s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/9fb7b446009e4e9aacb49a212a7d9e6f : SUCCESS in 38m 28s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/810724c63263430ca7487b837057af4e : SUCCESS in 37m 23s","accounts_in_message":[],"_revision_number":1},{"id":"f46babc724c4acc0159471f55d73659eaee73cdb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 17:13:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7da0ec89f16ed66ec33fc364d154469d9240f2a9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 18:50:15.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/80c7edd6fab74b98898ac7424220087e : SUCCESS in 10m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/818b571d27f94c3dbcb907dd5432dc2f : SUCCESS in 1h 14m 38s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/9025af3d41484305b2d4d60ae24b4eec : SUCCESS in 1h 26m 59s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/2ee510261c98490cabb2daf6b62c3f7e : SUCCESS in 1h 14m 13s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c116d3770ac64899841bee9f37fee346 : SUCCESS in 1h 15m 41s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/69c1d4e1ed6f4b67b88633cb80fb1126 : SUCCESS in 57m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/989f459f689d49ae84fcc2ec0e0bcae5 : SUCCESS in 51m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee6bb44b226748ccaae5aa24d33d52f0 : SUCCESS in 5m 18s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/045e3c33f853427eb89a62848c77424d : SUCCESS in 40m 45s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/73ce9b52b4594f1099c488c5e33b87dc : SUCCESS in 1h 09m 41s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/f51614b78bac41fb84635ed59eab960c : SUCCESS in 44m 42s","accounts_in_message":[],"_revision_number":1},{"id":"c776b6b120ef8c39aeeb6a54860b477633b1c1c3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 18:50:16.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9d568ced308f507aaf4bea35045dad3160145f89","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-05 18:51:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b58dcdbe3474d0fbf406401d14b91a3 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b4b2251f9b4fc61650188084b7071fbfe1d44de","revisions":{"8b4b2251f9b4fc61650188084b7071fbfe1d44de":{"kind":"REWORK","_number":1,"created":"2021-02-04 15:05:56.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/64/773964/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/64/773964/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/64/773964/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/64/773964/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/64/773964/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/64/773964/1"}}},"commit":{"parents":[{"commit":"3420f29a5557af0aa09b5aefe73b20de60f22cc7","subject":"Add no_log to LDAP domain config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/3420f29a5557af0aa09b5aefe73b20de60f22cc7"}]}],"author":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-02-03 17:09:27.000000000","tz":0},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-04 15:05:56.000000000","tz":0},"subject":"Allow OIDCClaimDelimiter to be set in the apache config file","message":"Allow OIDCClaimDelimiter to be set in the apache config file\n\nThis may be necessary for federation where there are multiple\nOIDC groups that are separate by a \u0027;\u0027. See [1].\n\n[1] https://docs.openstack.org/keystone/ussuri/admin/federation/mapping_combinations.html\n\nChange-Id: I68c0b138955693c8d1992f986878862ea12f5149\n(cherry picked from commit b71f4853e3749169f8d8e23c78c4dc3846c026dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8b4b2251f9b4fc61650188084b7071fbfe1d44de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/8b4b2251f9b4fc61650188084b7071fbfe1d44de"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
