)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~774788","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I09b181aabb904b976f6416ca7968aad4db7a1b09","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"wip_db_failure","hashtags":[],"change_id":"I09b181aabb904b976f6416ca7968aad4db7a1b09","subject":"DNM - Check reason for DB failure on focal","status":"ABANDONED","created":"2021-02-10 09:40:43.000000000","updated":"2021-03-19 03:57:02.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a10d25d35395410d8941c79318c98aa1a1d86ff1","_number":774788,"virtual_id_number":774788,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2021-02-10 13:11:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-10 13:11:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b5bb6133e869dc4088244f9a414975410aa5b212","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-10 09:40:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6c2702c4489de76bbb344e9eddcc5b71bf20c304","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-10 11:31:46.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"3430bfede2c03d53533f8d6537ec47822eff1b05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-10 13:11:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e908923461204f42adeb077b9f515205 : SUCCESS in 1h 07m 24s\n- openstack-ansible-deploy-aio_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4bab484472884daa9f49bf5a90d7747e : SUCCESS in 1h 33m 43s\n- openstack-ansible-deploy-aio_ovn_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/7631393cc75445de9a7935714a163bbd : SUCCESS in 1h 34m 13s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/1cac7048893e46548bba15db13d1b764 : SUCCESS in 54m 35s\n- openstack-ansible-deploy-aio_distro_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b6a4e2a29c2f4a4e8684557a732dbfc6 : FAILURE in 28m 42s\n- openstack-ansible-deploy-infra_lxc-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b7b0ae0180074455aa45ba710e2e3bbe : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":2},{"id":"a10d25d35395410d8941c79318c98aa1a1d86ff1","tag":"autogenerated:gerrit:abandon","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-19 03:57:02.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1cb4effdb2ea54af2a4f909937e1dcdd9ddc1ffc","revisions":{"2ded181042999dcc91bf70b1a6ebb10ce5e57f26":{"kind":"REWORK","_number":1,"created":"2021-02-10 09:40:43.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/774788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/88/774788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/88/774788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/88/774788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/88/774788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/88/774788/1"}}},"commit":{"parents":[{"commit":"20cf64fa0c1889d99e4cc1e2e46451d9cc7ffecd","subject":"Merge \"Move keystone pip package from a constraint to a requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/20cf64fa0c1889d99e4cc1e2e46451d9cc7ffecd"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-02-10 09:40:34.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-02-10 09:40:34.000000000","tz":120},"subject":"DNM - Check reason for DB failure on focal","message":"DNM - Check reason for DB failure on focal\n\nChange-Id: I09b181aabb904b976f6416ca7968aad4db7a1b09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2ded181042999dcc91bf70b1a6ebb10ce5e57f26"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/2ded181042999dcc91bf70b1a6ebb10ce5e57f26"}]},"branch":"refs/heads/master"},"1cb4effdb2ea54af2a4f909937e1dcdd9ddc1ffc":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2021-02-10 11:31:46.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/88/774788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/88/774788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/88/774788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/88/774788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/88/774788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/88/774788/2"}}},"commit":{"parents":[{"commit":"20cf64fa0c1889d99e4cc1e2e46451d9cc7ffecd","subject":"Merge \"Move keystone pip package from a constraint to a requirement\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/20cf64fa0c1889d99e4cc1e2e46451d9cc7ffecd"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-02-10 09:40:34.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-10 11:31:42.000000000","tz":0},"subject":"DNM - Check reason for DB failure on focal","message":"DNM - Check reason for DB failure on focal\n\n\nChange-Id: I09b181aabb904b976f6416ca7968aad4db7a1b09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1cb4effdb2ea54af2a4f909937e1dcdd9ddc1ffc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/1cb4effdb2ea54af2a4f909937e1dcdd9ddc1ffc"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
