)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~776181","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~stable%2Fvictoria~I08e784cf9505f92431176f66afc68c813a2334d9","project":"openstack/openstack-ansible-os_keystone","branch":"stable/victoria","hashtags":[],"change_id":"I08e784cf9505f92431176f66afc68c813a2334d9","subject":"Use new openstack.cloud collection names","status":"MERGED","created":"2021-02-17 19:31:36.000000000","updated":"2021-02-18 17:54:23.000000000","submitted":"2021-02-18 17:52:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"776181","cherry_pick_of_change":775923,"cherry_pick_of_patch_set":1,"meta_rev_id":"6442fd72cac6a8717ec088e2ceabf2b498aeebb8","_number":776181,"virtual_id_number":776181,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-02-18 17:52:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"all":[{"value":2,"date":"2021-02-18 08:22:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":2,"date":"2021-02-18 08:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":1,"date":"2021-02-18 08:41:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},{"value":0,"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"}],"CC":[{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-02-17 21:13:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-02-18 08:22:27.000000000","updated_by":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"reviewer":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"state":"REVIEWER"},{"updated":"2021-02-18 10:25:28.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"CC"}],"messages":[{"id":"58eaa0c5b1b07220ca195dc244f960fb3fb9a164","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-17 19:31:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e1c2ca36a8b1369f7c2bacf1c38cd0012d3cf4cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-17 21:13:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/dfc37441c56c442bb7a5fb18fb6e50be : SUCCESS in 11m 56s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a789da297d20492687c0bed9cf37acd3 : SUCCESS in 1h 04m 46s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/935eb43c578449aab325f09eb2353554 : SUCCESS in 1h 09m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1d5a3c44c0284fccaafe0feb10abbb3b : SUCCESS in 1h 11m 37s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eb6d54b73bbf46bc85e3612b632b0580 : SUCCESS in 1h 17m 58s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/26cb579928764c1a809a52134dd3d451 : SUCCESS in 1h 36m 09s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b87a1e798bd447c5ada3842b86875a11 : SUCCESS in 1h 27m 03s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ec0ece76617543f6b697c3f4b8b39bee : SUCCESS in 1h 08m 11s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/06ce4912fbcb4060a39bb2bb273e0cdd : SUCCESS in 52m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdde9662ba3248f8a30239acfc17a2d9 : SUCCESS in 7m 04s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/e7aac7ef818040de9abb5d08bfbcc1e0 : SUCCESS in 39m 48s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/71c9e4be25dd480ab947a089753d9b37 : SUCCESS in 42m 30s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a53efcf7902c4f3b9976d04f66928e29 : SUCCESS in 36m 07s","accounts_in_message":[],"_revision_number":1},{"id":"a4c61d0e07d0ef6bba01213add95942a05bd58f9","author":{"_account_id":31542,"name":"Andrew Bonney","email":"andrew.bonney@bbc.co.uk","username":"andrewbonney"},"date":"2021-02-18 08:22:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"f0ebb1d4fb3553a194a55b094264f17798e3cf84","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-02-18 08:41:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"937ce5a13d84e9913c650313abf01fa5b4cf5b8b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 08:41:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"664a7b2c08e66c08fd456ce6eb717e2e2568638d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 10:10:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3069ef70ce08444ea3033ecd956f1dfb : SUCCESS in 10m 03s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/6ef4be766a7e4644a06ed466b1529837 : SUCCESS in 1h 13m 31s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/a39a024d4e3a4372a8f5826f0336b1a4 : SUCCESS in 1h 20m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/b6917d912be34513834524c6b26d834e : SUCCESS in 1h 17m 02s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9e59ea3cb378410aa12667e03af227d7 : SUCCESS in 1h 03m 55s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6190879221984495ace26baa15984958 : SUCCESS in 57m 21s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/fe0abf769f3743cbaa922306b7f8e911 : SUCCESS in 52m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a086e0cee2d54adfa830aa3ff676ed2d : SUCCESS in 6m 49s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/2696b625c2f74c43a138614afc5027fc : FAILURE in 41m 23s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/56011b6c222146aa9eccb5c5d06c569b : FAILURE in 37m 21s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/359e741ebcb648a8bff8833cf99f2328 : FAILURE in 36m 07s","accounts_in_message":[],"_revision_number":1},{"id":"125d10e529871118295f002561ebca2a1fab168e","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-18 10:25:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"f205c4387dd5d1f45366944396ad283c96f53c7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 12:12:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ff2073de31e941b68b60970b8091bab8 : SUCCESS in 10m 31s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/4802852849a44c529e17cb98fe476417 : SUCCESS in 1h 03m 56s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/2a4c031250a7448c961dcb31ccd2ad8c : SUCCESS in 1h 09m 57s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bab3959d89ad40b8ab84effbfe1e133b : SUCCESS in 1h 07m 06s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5ee88c247e42473b9abedbac81463135 : SUCCESS in 1h 11m 42s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/328d4ee3877a4f8faa1033b45da8bd6e : SUCCESS in 1h 39m 24s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/4858d7dffabc4d32a1143e82b6e5ca7b : SUCCESS in 1h 26m 00s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/908bd698e50c48b5adcf18cb8ae35de6 : SUCCESS in 1h 06m 18s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/bc29e9747a294865b9e5d575bf30e153 : SUCCESS in 55m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2a2d8e793fd44208b77ab3b527bc46d0 : SUCCESS in 6m 11s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/647261b7e7eb42f8a15c07cb7cf1c869 : SUCCESS in 40m 28s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e7e90fa11aba479480271ec1ad5d4310 : TIMED_OUT in 1h 34m 23s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b2c6da0225c045f49106c804ec9006f2 : SUCCESS in 42m 05s","accounts_in_message":[],"_revision_number":1},{"id":"f0b4bba3be9139eba4c600e49d75354c21afdd38","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-02-18 14:22:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"d79a9d98d1b9b0a955765daa33a1cb73d9219db3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 16:27:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/6dd6c95ecc094584a2927102be7989f2 : SUCCESS in 10m 22s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/a035fcf4aa184ee6b578b8f0daa46393 : SUCCESS in 1h 09m 14s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/84a271bccf4640bd9785a341af7caff5 : SUCCESS in 1h 16m 26s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/12dc0842e9d04a54806ad9fbf956458c : SUCCESS in 1h 10m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/56c5b45ddd3f4fc6b78578f4897341ed : SUCCESS in 1h 12m 10s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/6334bec99cc340939fd39a5445ca4768 : SUCCESS in 1h 51m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/0d0c6a0e46a546219bdaa5cf04b1fb09 : SUCCESS in 1h 39m 54s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/229c96a2ab18402f8de2510373ac8200 : SUCCESS in 1h 08m 16s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ac6b70b22e0542e88757cea252f04a08 : SUCCESS in 46m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8cae0c24951460086fb670e103beaae : SUCCESS in 8m 43s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/ebe838d94a2242c4ad2a5aaaf3314cfd : SUCCESS in 43m 28s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/98ba9f702c71417b944666865580a9f7 : SUCCESS in 41m 28s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/43624a834f7a4d8895f1e7b9a2612b9f : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2141dc7a7a807e923ca117788516c2e9a4faab5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 16:27:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c7dfd715cef1b84001902116b5518f4412b98485","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:52:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/2a7d9a23dd0749ada0f04eeec424bc49 : SUCCESS in 9m 55s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/0586c8088d5b42bca2dfd5b7ccdd55b2 : SUCCESS in 53m 32s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/e1da92b734de4cfcad96d074f0898f55 : SUCCESS in 59m 52s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/1114db412bc348ea997f458166136aca : SUCCESS in 1h 11m 44s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/aa660b16ca2d4a3c89b289bd0056908a : SUCCESS in 1h 15m 29s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/77afe6205d604dedbe63a254b855f683 : SUCCESS in 58m 43s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/20a4f99014f746ffa070b5e1ae13b3c9 : SUCCESS in 54m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69019498453847a08a0602e163f57c04 : SUCCESS in 5m 58s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/7e5f25b3c38a499c9f9dc34c51d5360f : SUCCESS in 42m 13s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53661fd155cd4e8982d1eab1f14dc337 : SUCCESS in 43m 01s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5cc301a030054739bf66c8f2568b9af6 : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2d38ff23b9816c42536ef1195b5703a07bbe7dcd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:52:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6442fd72cac6a8717ec088e2ceabf2b498aeebb8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-02-18 17:54:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/737fc62a0e7b4187a010f961f8fa01b2 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a4bb13e134440b66ad316d41ef345e81939b1d9b","revisions":{"a4bb13e134440b66ad316d41ef345e81939b1d9b":{"kind":"REWORK","_number":1,"created":"2021-02-17 19:31:36.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/81/776181/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/81/776181/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/81/776181/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/81/776181/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/81/776181/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/81/776181/1"}}},"commit":{"parents":[{"commit":"ad079a5bade97424b69849ce675743f8440a518b","subject":"Use absolute path for uwsgi_params include","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/ad079a5bade97424b69849ce675743f8440a518b"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-02-16 19:50:00.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-02-17 19:31:36.000000000","tz":0},"subject":"Use new openstack.cloud collection names","message":"Use new openstack.cloud collection names\n\nos_ prefixed module names are deprecated for a while and will be removed\nCurrent usage generates a warning for users.\n\nChange-Id: I08e784cf9505f92431176f66afc68c813a2334d9\n(cherry picked from commit b30a49c5f8e1f320715c2c99c749047fb821046f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a4bb13e134440b66ad316d41ef345e81939b1d9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/a4bb13e134440b66ad316d41ef345e81939b1d9b"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
