)]}'
{"id":"openstack%2Fopenstack-ansible-os_keystone~781537","triplet_id":"openstack%2Fopenstack-ansible-os_keystone~master~I41ee59a901e54860067edd277fdade394b0b8858","project":"openstack/openstack-ansible-os_keystone","branch":"master","topic":"policy-json-to-yaml","hashtags":[],"change_id":"I41ee59a901e54860067edd277fdade394b0b8858","subject":"[goal] Deprecate the JSON formatted policy file","status":"MERGED","created":"2021-03-18 18:22:47.000000000","updated":"2021-04-09 10:05:07.000000000","submitted":"2021-04-09 10:03:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"781537-policy-json-to-yaml","meta_rev_id":"e4345b4ec8cebe7e57dfff6d1ec3087080235d1e","_number":781537,"virtual_id_number":781537,"owner":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-04-09 10:03:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":2,"date":"2021-04-09 08:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-04-06 13:41:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"all":[{"value":1,"date":"2021-04-09 08:23:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"}],"values":{"-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-03-18 20:50:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-06 13:41:46.000000000","updated_by":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"reviewer":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"state":"REVIEWER"},{"updated":"2021-04-09 08:23:11.000000000","updated_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"reviewer":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"state":"REVIEWER"}],"messages":[{"id":"8fe0dd76b31b54320523ba5de257a904e0391ddf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-18 18:22:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"832bf674d26d709778fd345651e914a6b33f5245","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-18 20:50:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/19cac4e1ed1d49258e330e1332267552 : SUCCESS in 9m 51s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/36bdd91817a24352aff303f0abd3c48c : SUCCESS in 1h 07m 59s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/df431fe7604f466f88fabc37a6633d41 : SUCCESS in 1h 09m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/53acd5892d2a49768a3c294afab3dba1 : SUCCESS in 1h 17m 32s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c0bfefc2586b4be7b12e6eb1d4816bb2 : SUCCESS in 1h 05m 37s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/887a25b4b33644059ce86108d772a992 : SUCCESS in 2h 15m 25s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/d7abdec1ff344fb89a7ef10bbac9391a : SUCCESS in 1h 50m 28s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/1a11290234484ee88b588411e4182589 : FAILURE in 43m 45s (non-voting)\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9ea842f53fd8425092b6ea44ff256fac : SUCCESS in 51m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e55df45f94524f2b8ab962d7493f0833 : SUCCESS in 11m 49s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/c2ac25d3c7bb41bf8b73876da7183e5f : FAILURE in 36m 35s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/f35c90bc6c8d4cd5b03d1e18839d5d55 : FAILURE in 43m 41s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5835bbf9f4bf4ffe86e1572c47b856a0 : FAILURE in 30m 33s","accounts_in_message":[],"_revision_number":1},{"id":"de91abefa903361262999db5f71c2d6fd61efa6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-03-24 12:13:54.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"aae68804b88ceef0c71901df8ef0d53ef9f2b822","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-24 14:12:11.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/3129183732d34082bdecc3a487cfc968 : SUCCESS in 9m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/c82325f724bf45ae807399ac3dccd76f : SUCCESS in 1h 12m 02s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b6d57b45b8644266af570b72bb0dedb7 : SUCCESS in 1h 13m 16s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cba5abeb71994caeb54956e038b0b495 : SUCCESS in 1h 27m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/54d507d8fd0341e6b4b2295b498d756c : SUCCESS in 1h 02m 11s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0029a0568ecd48d088339561d1f1619f : SUCCESS in 1h 48m 31s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/00c90d446a87458988f690cefc7f2a84 : SUCCESS in 1h 33m 30s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/7ba03f80e0e04f67a2eb341f88a5ed12 : SUCCESS in 56m 58s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/a6a2941bbe79471da6935eaa46c6b87a : SUCCESS in 41m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d33f78890a4157ab2845c2f77c359c : SUCCESS in 5m 23s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/6f6704bf6a1047daa62d907186fa0db9 : FAILURE in 39m 10s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/40d77a1b269d44d68de4721b16632951 : FAILURE in 32m 09s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/c5f76eac8a20446bb1bf7f5584f5e104 : FAILURE in 34m 20s","accounts_in_message":[],"_revision_number":2},{"id":"7f7e1b63cce1e103523cd29d6acb382dca03269f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-03-30 15:29:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"1b5c042abed46ae984550cfc4a870d7984226505","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-03-30 22:57:57.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/bd1744b0a5064b539cbae9954f060dfa : SUCCESS in 10m 35s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/d68d17eb55184f639d399083699a1858 : SUCCESS in 1h 13m 40s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/80d2303046014d2786438932414dd349 : SUCCESS in 1h 26m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/353bfea6b40a454e9641286953d792fc : SUCCESS in 1h 21m 29s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/e059bf37a9d446828fd1c8d489047ce2 : SUCCESS in 1h 15m 19s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/98fbd7d3576b4059b9e4f1cb81144728 : SUCCESS in 1h 50m 27s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/51155313abf24179840d64866489fef3 : SUCCESS in 1h 25m 02s\n- openstack-ansible-deploy-aio_distro_metal-centos-8 https://zuul.opendev.org/t/openstack/build/ece0040f9d2f40559fd0aac4530b2836 : FAILURE in 38m 57s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6f5b4a13bf994a51908eb55340b98349 : SUCCESS in 49m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40d6695a14d74063818a204d41ebd3b3 : SUCCESS in 6m 06s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/7cb108dacf14412d9e03a20ade8ca4ef : FAILURE in 10m 20s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/e9f0191f88034f0dab40a8e8c35a20c7 : FAILURE in 30m 13s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5e754a1590634901876a263170faa63a : FAILURE in 29m 26s","accounts_in_message":[],"_revision_number":3},{"id":"dd3d8c5d905294e7321c004793e2df4499694eb7","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-01 19:23:20.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"cbb511efc64a2c7152a83f288fead37aa4508329","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-01 21:33:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/78dd7ebfb8e9468d981047dbafbf765d : SUCCESS in 10m 38s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/e22f086ed9244225a5b0ac095413d4fe : SUCCESS in 1h 11m 11s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/0bac1413887d4e3eb2abf89c3fba66a2 : SUCCESS in 1h 19m 51s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/ba6b70def3294a49bf06605a50e9960f : SUCCESS in 1h 19m 23s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/6992bc65743d4fe8a0c51a049c38afaf : SUCCESS in 1h 10m 40s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/43de42bbfe1644ca826c58e4ab134095 : SUCCESS in 1h 47m 38s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/66f691285c344ede89d882be9b36f25f : SUCCESS in 1h 54m 23s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/6706fe81300d4fceb128d6f2d2059bbc : SUCCESS in 57m 22s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/ea2260ba67ca4df492c87e79f4e39873 : SUCCESS in 54m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/797292bbdd514598a39d2c405ec622b0 : SUCCESS in 7m 53s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/0cd29d8060bc4b379a60c972e00a9849 : FAILURE in 32m 03s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/02c3152505104061a2a479da09825671 : FAILURE in 30m 39s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/5d6c2dc1b6ba4b408a4318b6d47768d1 : FAILURE in 29m 14s","accounts_in_message":[],"_revision_number":3},{"id":"354c781204ce9976c0f45923efd4ef974e375e1e","author":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"date":"2021-04-05 12:36:46.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"fd8070bb355f0525bf536608e3661711d48bc6d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-05 14:26:01.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/ceb9d4d2d62e41c2bd56c94bd3a8e5b9 : SUCCESS in 10m 18s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/06723093b6d24dd7843b4fb16b1b6346 : SUCCESS in 1h 09m 15s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/c6b77346146c490a9e88180192a068dd : SUCCESS in 1h 05m 55s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/935d132866124c3bb9c30232b413f95d : SUCCESS in 1h 10m 17s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/91d3d7cea37e4a269d3f89ecd41fee9a : SUCCESS in 1h 06m 54s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/b8dfdd818b5e4d93be860c60fe2833c0 : SUCCESS in 1h 45m 02s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b4828faac24b4b0a8de4061741ba0dc8 : SUCCESS in 1h 36m 04s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/57d89d5f7f244130b35084e1aadf1b62 : SUCCESS in 58m 52s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/301cad64598c45d8af15ddf1f087d323 : SUCCESS in 58m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f516886ad6c450eabaabba1367a5404 : SUCCESS in 5m 26s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/67085b5340b54ce299d68f84ea5339ea : SUCCESS in 37m 55s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/bbf6c0ad75a440dc8f1a04ddb06f135c : SUCCESS in 35m 00s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/103a6cc057044decb48064e8fdedd6db : SUCCESS in 34m 15s","accounts_in_message":[],"_revision_number":3},{"id":"f10685bef95419901d3a441f0a5c2ce62acc57fa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"date":"2021-04-06 11:13:16.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a70f9b53b190c3cf171480cb97f582aa012f7fa5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-06 13:08:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/81a7aaf4fe2d4343a66086e44a1f4c3c : SUCCESS in 10m 48s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/87ebeceb51f044a8824b840dc089f6f7 : SUCCESS in 1h 10m 25s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/4cf9b3ec23684146b83c0962ac013410 : SUCCESS in 1h 10m 47s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/cbfbdc85582b4663af5e8bfbf4f18502 : SUCCESS in 1h 10m 53s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2f483d2a1d81436aab9901403851fd59 : SUCCESS in 1h 08m 31s\n- openstack-ansible-upgrade-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/f8fabcdde345433ca9772006161eebe6 : SUCCESS in 1h 46m 39s (non-voting)\n- openstack-ansible-upgrade-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/eaef5227e563416e8b89007c255ca198 : SUCCESS in 1h 33m 37s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/935e03b624c24531931f3e249a786655 : SUCCESS in 59m 59s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/2c7e6c6516f34ca7a6f17d27040db444 : SUCCESS in 1h 16m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/280539360d3d4d79bcab036a2e6ea767 : SUCCESS in 5m 39s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/c3fbba9686304eb2838e032d588a6b52 : SUCCESS in 48m 49s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/5a84dd1a6bbe4d72ae4db63430c7006f : SUCCESS in 41m 44s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/428367e5f6e344b09207ac9d8084092f : SUCCESS in 41m 10s","accounts_in_message":[],"_revision_number":4},{"id":"337ddecf85b3e56a6697344adc7011a36c6389c0","author":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"date":"2021-04-06 13:41:46.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"9dc6d7684df57ed6000f2773a58ac2860ba3cdb0","author":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"},"date":"2021-04-09 08:23:11.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"b73808407686e6be8f5611689ad67c4aa9ddc71b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 08:27:16.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"13323696744fd025d11e3ee822cff602662f0a91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 10:03:35.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-ansible-linters https://zuul.opendev.org/t/openstack/build/852d1f8e51444bc9adf7182ce9d6cfdd : SUCCESS in 10m 12s\n- openstack-ansible-deploy-aio_metal-debian-buster https://zuul.opendev.org/t/openstack/build/eb821f96f39d4b50bade798b47828c9d : SUCCESS in 1h 16m 43s\n- openstack-ansible-deploy-aio_metal-centos-8 https://zuul.opendev.org/t/openstack/build/02e954c8cf674a4796a69ec1631f0951 : SUCCESS in 1h 28m 14s\n- openstack-ansible-deploy-aio_metal-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/8023cd8add5f4ff380cd02fd8a37d277 : SUCCESS in 1h 11m 07s\n- openstack-ansible-deploy-aio_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/9c77b0a4ae7a414bb0ca45d02f3153df : SUCCESS in 1h 01m 12s\n- openstack-ansible-deploy-aio_distro_metal-centos-8-stream https://zuul.opendev.org/t/openstack/build/23e5ba5a243f4b46945015d375d369a9 : SUCCESS in 55m 12s\n- openstack-ansible-deploy-aio_distro_metal-ubuntu-focal https://zuul.opendev.org/t/openstack/build/228ec7dbe82043eca9e4ab23a2a415c6 : SUCCESS in 1h 03m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217f638892c24e28b60ca5391e75de91 : SUCCESS in 5m 37s\n- openstack-ansible-uw_apache-centos-8 https://zuul.opendev.org/t/openstack/build/771b028890cb40859da47013cd1f4bef : SUCCESS in 51m 34s\n- openstack-ansible-uw_apache-ubuntu-bionic https://zuul.opendev.org/t/openstack/build/28de7f0994d64473979d6b3ee53d38ef : SUCCESS in 38m 10s\n- openstack-ansible-uw_apache-ubuntu-focal https://zuul.opendev.org/t/openstack/build/b593a80092de489e83595471d3011af1 : SUCCESS in 46m 26s","accounts_in_message":[],"_revision_number":4},{"id":"c72864eb688513dd87b656dc74afb3191cb298dc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 10:03:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"e4345b4ec8cebe7e57dfff6d1ec3087080235d1e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-09 10:05:07.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e6a399e04c84803b084b6ff8227afec : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d89c6a153a251022027945f3469c394b52d532b8","revisions":{"7e65acaca322406a052a01fe7467bb1f6301700d":{"kind":"REWORK","_number":1,"created":"2021-03-18 18:22:47.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/781537/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/37/781537/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/1"}}},"commit":{"parents":[{"commit":"996d3c72bc90eafb6969ca5fc2a84c9f001c9fa0","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/996d3c72bc90eafb6969ca5fc2a84c9f001c9fa0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-18 18:22:36.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-18 18:22:36.000000000","tz":120},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to replace policy.json to\npolicy.yaml and remove deprecated policy.json.\n\nconfig_template has been choosen instead of the copy, since it can\nproperly handle content that has been lookuped.\n\nWe make a separate task not to restart service when it\u0027s not needed.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I41ee59a901e54860067edd277fdade394b0b8858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7e65acaca322406a052a01fe7467bb1f6301700d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7e65acaca322406a052a01fe7467bb1f6301700d"}]},"branch":"refs/heads/master"},"7e19fc73631fb44e63ea9bd630b992eba54dce00":{"kind":"REWORK","_number":2,"created":"2021-03-24 12:13:54.000000000","uploader":{"_account_id":28619,"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@gmail.com","username":"noonedeadpunk"},"ref":"refs/changes/37/781537/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/37/781537/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/2"}}},"commit":{"parents":[{"commit":"996d3c72bc90eafb6969ca5fc2a84c9f001c9fa0","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/996d3c72bc90eafb6969ca5fc2a84c9f001c9fa0"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-18 18:22:36.000000000","tz":120},"committer":{"name":"Dmitriy Rabotyagov","email":"noonedeadpunk@ya.ru","date":"2021-03-24 12:13:49.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to replace policy.json to\npolicy.yaml and remove deprecated policy.json.\n\nconfig_template has been choosen instead of the copy, since it can\nproperly handle content that has been lookuped.\n\nWe make a separate task not to restart service when it\u0027s not needed.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I41ee59a901e54860067edd277fdade394b0b8858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7e19fc73631fb44e63ea9bd630b992eba54dce00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/7e19fc73631fb44e63ea9bd630b992eba54dce00"}]},"branch":"refs/heads/master"},"6dd88a458253ad662433a4a7e3537af8c3ed6fe4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2021-03-30 15:29:01.000000000","uploader":{"_account_id":25023,"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","username":"jrosser"},"ref":"refs/changes/37/781537/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/37/781537/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/3"}}},"commit":{"parents":[{"commit":"054f7b545cb282c3ac5d07d65dfc2463f913ee96","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/054f7b545cb282c3ac5d07d65dfc2463f913ee96"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-18 18:22:36.000000000","tz":120},"committer":{"name":"Jonathan Rosser","email":"jonathan.rosser@rd.bbc.co.uk","date":"2021-03-30 15:29:01.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to replace policy.json to\npolicy.yaml and remove deprecated policy.json.\n\nconfig_template has been choosen instead of the copy, since it can\nproperly handle content that has been lookuped.\n\nWe make a separate task not to restart service when it\u0027s not needed.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I41ee59a901e54860067edd277fdade394b0b8858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6dd88a458253ad662433a4a7e3537af8c3ed6fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/6dd88a458253ad662433a4a7e3537af8c3ed6fe4"}]},"branch":"refs/heads/master","description":"Rebase"},"d89c6a153a251022027945f3469c394b52d532b8":{"kind":"REWORK","_number":4,"created":"2021-04-06 11:13:16.000000000","uploader":{"_account_id":29865,"name":"Georgina Shippey","email":"georgina.shippey@outlook.com","username":"gshippey"},"ref":"refs/changes/37/781537/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/openstack-ansible-os_keystone","ref":"refs/changes/37/781537/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/openstack-ansible-os_keystone refs/changes/37/781537/4"}}},"commit":{"parents":[{"commit":"054f7b545cb282c3ac5d07d65dfc2463f913ee96","subject":"Updated from OpenStack Ansible Tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/054f7b545cb282c3ac5d07d65dfc2463f913ee96"}]}],"author":{"name":"Dmitriy Rabotyagov","email":"dmitriy.rabotyagov@citynetwork.eu","date":"2021-03-18 18:22:36.000000000","tz":120},"committer":{"name":"Georgina Shippey","email":"georgina.shippey@bbc.co.uk","date":"2021-04-06 11:13:10.000000000","tz":0},"subject":"[goal] Deprecate the JSON formatted policy file","message":"[goal] Deprecate the JSON formatted policy file\n\nAs per the community goal of migrating the policy file\nthe format from JSON to YAML[1], we need to replace policy.json to\npolicy.yaml and remove deprecated policy.json.\n\nconfig_template has been choosen instead of the copy, since it can\nproperly handle content that has been lookuped.\n\nWe make a separate task not to restart service when it\u0027s not needed.\n\n[1] https://governance.openstack.org/tc/goals/selected/wallaby/migrate-policy-format-from-json-to-yaml.html\n\nChange-Id: I41ee59a901e54860067edd277fdade394b0b8858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d89c6a153a251022027945f3469c394b52d532b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/openstack-ansible-os_keystone/commit/d89c6a153a251022027945f3469c394b52d532b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":25600,"name":"Ebbex","display_name":"ebbex","email":"openstack@slipsprogrammor.no","username":"ebbex"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
